)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~951135","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~Ife55fa203fd5bdef7e9bc4b8b3f5273d3dc7bfa6","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"pip-23.1-support","hashtags":[],"change_id":"Ife55fa203fd5bdef7e9bc4b8b3f5273d3dc7bfa6","subject":"add pyproject.toml to support pip 23.1","status":"MERGED","created":"2025-05-28 13:15:14.000000000","updated":"2025-05-29 05:14:25.000000000","submitted":"2025-05-29 05:14:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951135-pip-23.1-support","meta_rev_id":"8845a534af4fcf5459234fd887c2e4996670f08c","_number":951135,"virtual_id_number":951135,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-29 05:14:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-05-29 04:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-05-29 04:29:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-28 14:23:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab39bc7bcd75718419ea2b6126169c724af9e4d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-28 13:15:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb7e311e613d1d7ffa237be91056bfa702571d9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 14:23:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fafdd052ce3b4d57b68fd5ff7a721b9e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/191bc364c7dd4ac0b62da1fa43f2a0ee : SUCCESS in 2m 21s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/a49496f12c4748aa888a8466eecef8cc : SUCCESS in 31m 25s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/acd064f54b8a4186b1e74c5be467c909 : SUCCESS in 32m 39s\n- zaqar-tempest-plugin-mongodb-2023-2 https://zuul.opendev.org/t/openstack/build/e8bbaded46e24e8bb5a6d10c5a610aaf : SUCCESS in 28m 08s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/92d9de2187284cebb20788996be20840 : SUCCESS in 29m 31s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/b3ddaeb7120c41908a5394e148adade0 : SUCCESS in 30m 05s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/747c259996874aabac9fc5d146e8f229 : SUCCESS in 29m 23s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/0b97fb75788445d09a67132d2229ba89 : SUCCESS in 29m 12s\n- zaqar-tempest-plugin-redis-2023-2 https://zuul.opendev.org/t/openstack/build/dbb6c6f183d54c6a835e81a1a6ded485 : SUCCESS in 27m 46s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/65ae43e7de624bc9817870d9c0411d60 : SUCCESS in 26m 53s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/f59b076c68f649f685a4bbdaf1aa1e7f : SUCCESS in 22m 18s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/ec63d894b294465c8b28dfedac4b7bd9 : SUCCESS in 25m 22s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/bbd80094b6a3449cafbe3f25955b9ba4 : SUCCESS in 30m 58s\n- zaqar-tempest-plugin-swift-2023-2 https://zuul.opendev.org/t/openstack/build/295a9932b89040548f30b24d0049baea : SUCCESS in 29m 30s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/c5805add9c354e5a8fa85cc05241dced : SUCCESS in 33m 23s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/5dbb4206d10647f9a78b44678b604b55 : SUCCESS in 28m 07s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/5204705fed734587bd868a246b0865eb : SUCCESS in 33m 00s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/bb344d190ec04e55ab676d35c3fe1eb6 : SUCCESS in 27m 39s","accounts_in_message":[],"_revision_number":1},{"id":"69201eeeae3cbfe6885445e6849676abb49d9e43","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-29 04:29:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4cc4e2fd0e0c959b513fe8516ad028ff42ee4b1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 04:29:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec192d99d1fd107979a1820507c2292f9e0b2ff4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 05:14:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aec206067ca42dc8ca592d5022f402b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e8ac7d3fe5d4785a210f4d617a8547f : SUCCESS in 2m 06s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/515870062e8c4989845b5066dd0bdbfa : SUCCESS in 24m 05s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/cdaf099ad79b4ba6adac726209c1e92c : SUCCESS in 23m 58s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/5ae3651fb93548129790c89af29653e0 : SUCCESS in 31m 26s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/88dfd18542444a8a89313ab5968bcd59 : SUCCESS in 32m 56s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/bd0ff582aeff475e8879949729f86cd4 : SUCCESS in 26m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8845a534af4fcf5459234fd887c2e4996670f08c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 05:14:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73d8fe296002ee32dcd69dacfc65304a28588e2f","revisions":{"73d8fe296002ee32dcd69dacfc65304a28588e2f":{"kind":"REWORK","_number":1,"created":"2025-05-28 13:15:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/951135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/35/951135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/35/951135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/35/951135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/35/951135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/35/951135/1"}}},"commit":{"parents":[{"commit":"a7504dfc3df5960c2f8a0c1852e9acd8e63ea921","subject":"Drop redundant injection of VIRTUAL_ENV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/a7504dfc3df5960c2f8a0c1852e9acd8e63ea921"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-28 13:14:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-28 13:14:50.000000000","tz":540},"subject":"add pyproject.toml to support pip 23.1","message":"add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nCo-Authored-By: Sean Mooney \u003cwork@seanmooney.info\u003e\nChange-Id: Ife55fa203fd5bdef7e9bc4b8b3f5273d3dc7bfa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/73d8fe296002ee32dcd69dacfc65304a28588e2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/73d8fe296002ee32dcd69dacfc65304a28588e2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
