)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~951228","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I8aae03e9418fd52fe0da876501e613367e9d5027","project":"openstack/zaqar-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-29 05:44:51.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I8aae03e9418fd52fe0da876501e613367e9d5027","subject":"Remove 2023.2 job and jammy job","status":"MERGED","created":"2025-05-29 05:22:23.000000000","updated":"2025-05-29 09:28:46.000000000","submitted":"2025-05-29 09:28:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951228","meta_rev_id":"d5d29e430b069159d64148e4f20eeeb7b7273f7e","_number":951228,"virtual_id_number":951228,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-29 09:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-05-29 08:34:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-05-29 08:34:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-29 05:23:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9ab45a9547e09c389d63a3d4ed2384921359e08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-29 05:22:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d342ad98b560ecf262b8272ad71e44344d7617d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 05:23:09.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/zaqar on branch stable/2025.1.  The\nproblem was:\n\n  Job zaqar-tempest-plugin-mongodb-jammy not defined\n\nThe problem appears in the the \"opendev.org/openstack/zaqar\" project stanza:\n\n  project:\n      queue: zaqar\n      templates:\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n  ...\n\n  in \"openstack/zaqar/.zuul.yaml@stable/2025.1\", line 76","accounts_in_message":[],"_revision_number":1},{"id":"8ecb33b3d5eb47e965854a6e48ffda1b1fe38a04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-29 05:39:35.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8053fde5ac6773efbb46037ace017c8d9ab49c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 05:39:46.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/zaqar on branch stable/2025.1.  The\nproblem was:\n\n  Job zaqar-tempest-plugin-mongodb-jammy not defined\n\nThe problem appears in the the \"opendev.org/openstack/zaqar\" project stanza:\n\n  project:\n      queue: zaqar\n      templates:\n        - openstack-python3-jobs\n        - openstack-python3-jobs-arm64\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n  ...\n\n  in \"openstack/zaqar/.zuul.yaml@stable/2025.1\", line 76","accounts_in_message":[],"_revision_number":2},{"id":"4bc90a217ba42d91ee88457ff7031e5542058969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-29 05:44:29.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5d3de7bc3ce7e0c8ff5a095f13395e005f9b1ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 07:02:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6e4709300d247799adb57a1dcb1d1e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c848356a472741e8a2cc3d6864ca9e8c : SUCCESS in 2m 09s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/6910ad576342445fb0219e9404eee760 : SUCCESS in 31m 56s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/0477b0a45c27464488287029069fd18e : SUCCESS in 23m 51s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/025d0ff8f7a046c29e450003d5d41806 : SUCCESS in 22m 15s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/d86d941d26bd46a3ab0959dfac159c13 : SUCCESS in 30m 46s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/ba8e53c31f3f4ecd8d4a91c62ab3390f : SUCCESS in 27m 09s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/172014724d4a40b7ba202dd973eb26c7 : SUCCESS in 28m 08s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/49d9807e744e456a9cc7c49bfac7a7a9 : SUCCESS in 25m 21s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/b12b50eba5284071a4abd45339460e67 : SUCCESS in 33m 00s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/123a48bea7b44d9ebac9c7cf942fe0d4 : SUCCESS in 25m 12s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/dc37d088540d4f41ae5642c4f461735c : SUCCESS in 37m 45s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/0865d6b49846452992ef752e964f3995 : SUCCESS in 28m 06s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/4a92b4f0a99f42808f98d1376e790c14 : SUCCESS in 34m 55s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/a27afe6297544b628b4c5fa39e62aad1 : SUCCESS in 25m 11s","accounts_in_message":[],"_revision_number":3},{"id":"b14ef35ecc6c18f495a686fda5a8ea1d0cf37c74","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-29 08:34:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5ae66acb2f43e43f81008c179f71e09317cfa86b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 08:34:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c4c2e5e2eff11e19005fc7912e95f0532d23ffa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 09:28:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb7be7ea2314476080eaa322d5624beb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c08a83e3833b4e77a493bf91a67100a3 : SUCCESS in 2m 14s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/f1258633c05d4650b9885760aaf5e043 : SUCCESS in 26m 30s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/eb101b41464e46b59cb195a61bc719ec : SUCCESS in 23m 59s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/481e56672a8340b0949ec72541974a66 : SUCCESS in 26m 11s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/6a4dddde556e46c79d777a85c61cc187 : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":3},{"id":"d5d29e430b069159d64148e4f20eeeb7b7273f7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 09:28:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0c98f7531014ee4b8b7b272569ec160f31ac8336","revisions":{"f3ad296937ecd6431caf4d5e67e2662d2ed0b440":{"kind":"REWORK","_number":1,"created":"2025-05-29 05:22:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/951228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/28/951228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/1"}}},"commit":{"parents":[{"commit":"73d8fe296002ee32dcd69dacfc65304a28588e2f","subject":"add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/73d8fe296002ee32dcd69dacfc65304a28588e2f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 05:21:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 05:21:44.000000000","tz":540},"subject":"Remove 2023.2 job and jammy job","message":"Remove 2023.2 job and jammy job\n\n2023.2 already reached its EOL. Also Ubuntu Jammy is no longer\nsupported by the latest master.\n\nChange-Id: I8aae03e9418fd52fe0da876501e613367e9d5027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/f3ad296937ecd6431caf4d5e67e2662d2ed0b440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/f3ad296937ecd6431caf4d5e67e2662d2ed0b440"}]},"branch":"refs/heads/master"},"f67bf8be9b163b32f82947fd30c4b15930971bdd":{"kind":"REWORK","_number":2,"created":"2025-05-29 05:39:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/951228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/28/951228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/2"}}},"commit":{"parents":[{"commit":"73d8fe296002ee32dcd69dacfc65304a28588e2f","subject":"add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/73d8fe296002ee32dcd69dacfc65304a28588e2f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 05:21:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 05:39:35.000000000","tz":0},"subject":"Remove 2023.2 job and jammy job","message":"Remove 2023.2 job and jammy job\n\n2023.2 already reached its EOL. Also Ubuntu Jammy is no longer\nsupported by the latest master.\n\nChange-Id: I8aae03e9418fd52fe0da876501e613367e9d5027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/f67bf8be9b163b32f82947fd30c4b15930971bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/f67bf8be9b163b32f82947fd30c4b15930971bdd"}]},"branch":"refs/heads/master"},"0c98f7531014ee4b8b7b272569ec160f31ac8336":{"kind":"REWORK","_number":3,"created":"2025-05-29 05:44:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/951228/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/28/951228/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/28/951228/3"}}},"commit":{"parents":[{"commit":"73d8fe296002ee32dcd69dacfc65304a28588e2f","subject":"add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/73d8fe296002ee32dcd69dacfc65304a28588e2f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 05:21:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-29 05:44:29.000000000","tz":0},"subject":"Remove 2023.2 job and jammy job","message":"Remove 2023.2 job and jammy job\n\n2023.2 already reached its EOL. Also Ubuntu Jammy is no longer\nsupported by the latest master.\n\nChange-Id: I8aae03e9418fd52fe0da876501e613367e9d5027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/0c98f7531014ee4b8b7b272569ec160f31ac8336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/0c98f7531014ee4b8b7b272569ec160f31ac8336"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
