)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~956438","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I4a0734a9ec878f75f2b47d31c2b6a30b6e63a333","project":"openstack/zaqar-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-11 18:09:54.000000000","reason":"Change was submitted"},"8846":{"account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"last_update":"2025-08-11 17:21:29.000000000","reason":"\u003cGERRIT_ACCOUNT_8846\u003e replied on the change","reason_account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}}},"hashtags":[],"change_id":"I4a0734a9ec878f75f2b47d31c2b6a30b6e63a333","subject":"Remove remaining client for v1 API","status":"MERGED","created":"2025-08-03 09:08:01.000000000","updated":"2025-08-11 18:09:54.000000000","submitted":"2025-08-11 18:09:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956438","meta_rev_id":"19ce5585701b4746bb32f99cce9e5e5a96d6bddc","_number":956438,"virtual_id_number":956438,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-11 18:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-11 17:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-11 17:21:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-03 09:51:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-04 15:15:08.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"REVIEWER"}],"messages":[{"id":"f58b7dcacd72e2ead87584521177c721c7b7b1e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-03 09:08:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"facc8394f53d19350070e7dfcfd40ae9046d7377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 09:51:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e3d5e5af3af44c297c197054dea0775\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28af165608734c39aa82ca36f5798fbf : FAILURE in 2m 38s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/7a4b2fd413db45c4a73a7167c0ae56d8 : SUCCESS in 30m 20s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/c7be91dade164f7b82471b419d07bd05 : SUCCESS in 33m 58s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/835f29f53f444357ab1e585e42eab1a2 : SUCCESS in 31m 39s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/902343b3874f404f9f9b6776c471241d : FAILURE in 24m 59s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/8d3f766477714b32a807490b9fac9f5d : SUCCESS in 36m 55s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/ad76eb79685c4a25b408e0b3189b02d7 : SUCCESS in 29m 18s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/3f9f5ab0858d4410a7f3b7c3620e0b20 : SUCCESS in 27m 54s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/f312378919ee438ba5a3621b0b56c3ec : SUCCESS in 32m 04s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/01d3402d339a45a2b78bac9bc4411282 : SUCCESS in 30m 19s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/8181b91af2c34a529188725fa3f71f0d : SUCCESS in 21m 02s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/ccceb748bf53475bb14bbe840ae75e83 : SUCCESS in 28m 45s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/ba96716a21da40e6a79fc4324025cb91 : SUCCESS in 21m 31s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/9db2699655014a8d94255c49a231803f : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":1},{"id":"704dc4aeeb123487bd3a7fe69f73783f73c4cb54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-03 14:38:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8f161ac499cab7d0336a8993823336cc25eff650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 15:19:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c05926fce17e49d19a004f08646272fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ae7a16376e407fb092da5dccf76c72 : SUCCESS in 2m 17s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/67dc0982fbc7404aa5107c0d04890cc3 : SUCCESS in 30m 09s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/eb1b1e01c9d040babcb7b6ffe80bf816 : SUCCESS in 32m 39s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/1437874e757b4e74b658cc88d30a2cda : SUCCESS in 15m 48s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/91dfb38fe3544ea88c151f050bca76c5 : FAILURE in 9m 53s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/52a7c981681840e194300cc298d99693 : SUCCESS in 15m 36s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/2e42473b732c41518d3f0424a12b65d6 : SUCCESS in 23m 57s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/22fbce6f8a514184a90bcf02b62918a6 : SUCCESS in 21m 11s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/b6fdd9981beb435a909c1bc97fccfffb : SUCCESS in 15m 14s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/01c69a71be7c4a30b55b61ef3ac34f15 : SUCCESS in 15m 36s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/f7f38139673647a68dde7b810f75063f : SUCCESS in 34m 30s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/e1b674baf092413f98908dc8829eb045 : SUCCESS in 35m 24s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/3cebb4993439405180cf64de1679a576 : SUCCESS in 32m 44s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/5024640be8fe4b14b01845e278365be7 : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":2},{"id":"29e5d4a4f9558fb3cd7ea192e7bee35e9ef2e307","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-03 15:20:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"361e0f58aab26893de5c3f268df17e61b97b4369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 16:09:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10d6f8ede48440828b57543df0979ae0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9945556f0a564aa793f145922e97ae52 : SUCCESS in 2m 37s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/3da56e63be42470c9468a900fc39b6fb : SUCCESS in 14m 53s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/ae7c08f68dfd40b08f633fe03920f6d8 : SUCCESS in 18m 49s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/6313dfbe1cbc434296ce427e871cf58a : SUCCESS in 13m 43s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/bba999bae6d24ed49b476db3d7df0564 : FAILURE in 15m 59s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/7a260a81ded84a4db732b3e65ba591fc : SUCCESS in 30m 26s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/36077f0d636a42f0b04beef9a5aaaf6b : SUCCESS in 19m 42s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/99ff7f4754c649be9fc379b25a85a469 : SUCCESS in 15m 55s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/1f6fa5531d0741e59fa396567ecb1a76 : SUCCESS in 33m 41s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/48b6f0bcb74c4aa1a31aaf7ddc1da0c2 : SUCCESS in 31m 29s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/881ad0c9452d49c4b4d8c3c1d4690910 : SUCCESS in 22m 34s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/c12708c88cd645a08b89ba50ff6c617b : SUCCESS in 28m 39s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/c5fa974a185b4728b78264c6dae157fe : SUCCESS in 41m 20s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/9c5f442eb9744bad99f197cfd2fad541 : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":2},{"id":"373ea97bc092bcf2e33d523720082eba54d0ce86","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-04 06:27:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f809f86f18ef21430cfd1e02e863c7cc7c1316e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 07:05:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/469939c38235477cb1cd82211f0e3fc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c61b1110e9a9473abf14efad5c0284f0 : SUCCESS in 3m 35s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/2d6e18c8d06548898ff2bcf36605fdd2 : SUCCESS in 29m 18s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/218b5e9f1d7442b5aea130f1bfcfc8b8 : SUCCESS in 14m 21s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/6325375d4331433a9174391d38e27c5b : SUCCESS in 26m 47s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/83a6f18356684edeb9b27d35b5ab9090 : SUCCESS in 31m 54s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/4dca4758033a483e93da1b34053afe2a : SUCCESS in 29m 29s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/0e9b3857317541f1ac1ee6989e14b484 : SUCCESS in 20m 59s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/f3080dd831a3490b824e49d38ba6f199 : SUCCESS in 26m 05s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/09b807add55b4838a39d66732c44078c : SUCCESS in 23m 16s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/debc216e2fad46808e6a965e9df6e58b : SUCCESS in 31m 47s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/d4e53e63c0e5435584f2ac661e45cfe4 : SUCCESS in 31m 31s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/0f50f757fee54232955336565cee80df : SUCCESS in 28m 05s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/85acb8aca63f4361b9c3d55728e77596 : SUCCESS in 31m 30s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/246b0a5b13ff45ec814bdda14b1b0781 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f3be6fe1cf05f37de195416611af842bbf22a1e8","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-08-11 17:21:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd91e183632a2e4f0609b554ebd7c93a61cbddac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 17:22:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"642ed6e85096146bff53043814bdba659db01e54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 18:09:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f1daae052a24451bf844b2351f5a6e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b545caace975442abf9e64f09597715e : SUCCESS in 2m 14s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/14396cbe5d734f36ac405c5c67ca2860 : SUCCESS in 31m 34s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/096c5ef010334182adff6a6fbca56205 : SUCCESS in 36m 14s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/4c56a8836024407e981d38424685e0b4 : SUCCESS in 34m 43s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/ea10cf12897344b1bcd0df16d8d45321 : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":2},{"id":"19ce5585701b4746bb32f99cce9e5e5a96d6bddc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 18:09:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce820031e72df551889334f98132683be42c713c","revisions":{"72db159356bc574bf2f10ca91e9e75c56a3bfe37":{"kind":"REWORK","_number":1,"created":"2025-08-03 09:08:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/956438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/38/956438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/956438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/956438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/956438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/956438/1"}}},"commit":{"parents":[{"commit":"73e2764d7d2a013393dbc20e59888b648e133289","subject":"Drop explicit dependency on python-subunit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/73e2764d7d2a013393dbc20e59888b648e133289"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-03 09:07:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-03 09:07:52.000000000","tz":540},"subject":"Remove remaining client for v1 API","message":"Remove remaining client for v1 API\n\n... because tests for V1 API were already removed.\n\nAlso refactor how version and uri prefix are initialized.\n\nChange-Id: I4a0734a9ec878f75f2b47d31c2b6a30b6e63a333\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/72db159356bc574bf2f10ca91e9e75c56a3bfe37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/72db159356bc574bf2f10ca91e9e75c56a3bfe37"}]},"branch":"refs/heads/master"},"ce820031e72df551889334f98132683be42c713c":{"kind":"REWORK","_number":2,"created":"2025-08-03 14:38:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/956438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/38/956438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/956438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/956438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/956438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/38/956438/2"}}},"commit":{"parents":[{"commit":"73e2764d7d2a013393dbc20e59888b648e133289","subject":"Drop explicit dependency on python-subunit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/73e2764d7d2a013393dbc20e59888b648e133289"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-03 09:07:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-03 14:38:12.000000000","tz":540},"subject":"Remove remaining client for v1 API","message":"Remove remaining client for v1 API\n\n... because tests for V1 API were already removed.\n\nAlso refactor how version and uri prefix are initialized.\n\nChange-Id: I4a0734a9ec878f75f2b47d31c2b6a30b6e63a333\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/ce820031e72df551889334f98132683be42c713c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/ce820031e72df551889334f98132683be42c713c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
