)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~960830","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~Ie60c34673bac9b2cef6d26199c0ffd171ffe0cd4","project":"openstack/zaqar-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ie60c34673bac9b2cef6d26199c0ffd171ffe0cd4","subject":"Add stable/2025.2 job","status":"MERGED","created":"2025-09-12 13:57:02.000000000","updated":"2025-09-12 15:45:05.000000000","submitted":"2025-09-12 15:45:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960830","meta_rev_id":"36e2526d5d4f564b7eff68094aba13fab0e5e8da","_number":960830,"virtual_id_number":960830,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-12 15:45:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-09-12 15:04:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-09-12 15:04:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-12 15:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2de4e26e966a3ab6350537839f58c69263b2e4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-12 13:57:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b6f50786f660ba6163b1c14f233e71c3e808bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 15:04:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf59dab900d4db384acc92c1a711891\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b242d4481374b43831432e3b1c33333 : SUCCESS in 2m 55s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/2d7b8625b0ba4c85ad7577f6371dc8bc : SUCCESS in 34m 22s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/5710d908a53b454fa15e7e84aa156c0a : SUCCESS in 31m 21s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/ea2fd70f42a4499690ec079f5b645580 : SUCCESS in 22m 01s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/801427dd4d5c45d5be97236c76b6b3e4 : SUCCESS in 31m 41s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/82d11663624b4b9d84a3212c08c3ef6b : SUCCESS in 14m 58s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/4c09fdb3e40844be91a492bbdd9eaf0e : SUCCESS in 33m 48s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/dbe373e9db9c4cbaa9239c1180abb9e0 : SUCCESS in 30m 08s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/3965d61435e9434690048a2b9cca202a : SUCCESS in 30m 21s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/299d035b7e8244b6968d00ec2375eb90 : SUCCESS in 35m 10s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/28dd7dc5a78341cb934c8e6928d46e34 : SUCCESS in 17m 49s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/0bd2131b076547c898a2e2dd99461870 : SUCCESS in 35m 34s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/845172fdcdca42e5b51ac96b77d7f973 : SUCCESS in 33m 32s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/db299817240843108b3f726467443246 : SUCCESS in 22m 57s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/8e7b2705a5dc4166937f0986292715b8 : SUCCESS in 58m 39s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/c9ecf9f3425949d694fc780997ffc58b : SUCCESS in 34m 13s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/2f21e81c807646229b0a74be1be8a279 : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":1},{"id":"92eaac0ac7b1e8774da8cc080ec7afa4499db048","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-12 15:04:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c866fa315df8f76b6195503ffc6fbbd01087207","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 15:05:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c05cb8a4c528bc5ec79c2338d509b19cc31aa388","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 15:45:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b1a0544452c4264801bebd7259ab334\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35ee697ffd6d4f50b04181cc1e236ab9 : SUCCESS in 2m 29s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/9f9e942e5ac9478aaca602b2cdd159b6 : SUCCESS in 33m 05s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/7c59f83094be41e8969e9580f9519d36 : SUCCESS in 32m 23s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/9f012b3ae70942b2bf2f27a1d54d0dfc : SUCCESS in 35m 23s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/a02019f5ab2d4412ae827c98b2a0f34a : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":1},{"id":"36e2526d5d4f564b7eff68094aba13fab0e5e8da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 15:45:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5c3246767cd4cf31d84aab9cfb3dee3520295bf","revisions":{"b5c3246767cd4cf31d84aab9cfb3dee3520295bf":{"kind":"REWORK","_number":1,"created":"2025-09-12 13:57:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/960830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/30/960830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/30/960830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/30/960830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/30/960830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/30/960830/1"}}},"commit":{"parents":[{"commit":"ce820031e72df551889334f98132683be42c713c","subject":"Remove remaining client for v1 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/ce820031e72df551889334f98132683be42c713c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-12 13:56:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-12 13:56:54.000000000","tz":540},"subject":"Add stable/2025.2 job","message":"Add stable/2025.2 job\n\nChange-Id: Ie60c34673bac9b2cef6d26199c0ffd171ffe0cd4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/b5c3246767cd4cf31d84aab9cfb3dee3520295bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/b5c3246767cd4cf31d84aab9cfb3dee3520295bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
