)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~962231","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I8f8045205b0fa1832118cd06fd1a1e0a9b81e96f","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"remove-api-v1.1-support","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-07 19:14:24.000000000","reason":"Change was submitted"},"8846":{"account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"last_update":"2025-09-26 16:51:43.000000000","reason":"\u003cGERRIT_ACCOUNT_8846\u003e replied on the change","reason_account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}}},"hashtags":[],"change_id":"I8f8045205b0fa1832118cd06fd1a1e0a9b81e96f","subject":"Remove api v1.1 support in tempest plugin","status":"MERGED","created":"2025-09-24 21:25:32.000000000","updated":"2025-10-07 19:14:24.000000000","submitted":"2025-10-07 19:14:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962231-remove-api-v1.1-support","meta_rev_id":"92020ad67580c15c40cb57d882f42087741457f8","_number":962231,"virtual_id_number":962231,"owner":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-07 19:14:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-07 18:38:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-07 18:38:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-24 21:26:33.000000000","updated_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-09-24 22:01:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89fcaf65555ecd75ec2ae1148fcd175568adac9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-09-24 21:25:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4530d5c5c18807d8183add7571b3577549b3bbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 22:01:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29d37d3259e34ad196da0945f1c3b86b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a37db8c1a8cb4d25895d2d93b40604f0 : SUCCESS in 1m 44s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/01a36803639d4b4a90316486c83d786d : FAILURE in 27m 19s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/76681754e1354a599c2c394615760a09 : FAILURE in 12m 12s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/cf5fde4017aa413e9a40527a6907d2e3 : FAILURE in 12m 52s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/8e3f461ff2ca422b838d7bb0d0360222 : FAILURE in 14m 24s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/fe638f5d8f474890b161539523c9e50c : FAILURE in 26m 57s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/78224fc3542d4d419478c97d415d2326 : FAILURE in 14m 49s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/1606c229cd464d95bb23ee5b6fd9ad3e : FAILURE in 27m 30s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/035275976bbd473597d5c25c2b6ae564 : FAILURE in 24m 23s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/43e70b788ac842939bd04e523350344f : FAILURE in 22m 27s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/e3415278ebbe4f0e8ab47b73d065c1ba : FAILURE in 30m 41s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/2503db452ae94d5e8f9635f992d6008b : FAILURE in 29m 16s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/443cb3491ac74754a60067231dbf91ce : FAILURE in 16m 09s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/71a9091af89148a09ff8e64e9fb86cb1 : FAILURE in 15m 19s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/86fcfc448de24170a46782c0d21270c4 : FAILURE in 16m 38s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/d9aef2f38f594bbcbf67687c5b70566f : FAILURE in 29m 30s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/d8fac9d5bb504d8cb86512278ee953b8 : FAILURE in 29m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9025a7a468d9b501ac58ebfe04e3d4d4efc175bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-09-25 17:31:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1d907a3202d3b534a3407d99f3ab99a12bf71b6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 18:10:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a7d0e054265430cb0dbce07b7113591\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0e9594d2a34c9d83137640079a06ac : SUCCESS in 2m 03s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/49691fd0df5a41ca9c523e1b1da636af : SUCCESS in 21m 48s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/97b49ec65d8f46549af04066cba9d40b : SUCCESS in 22m 35s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/3b66e2a93d05410b90fe81bf86437c7f : SUCCESS in 27m 43s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/96446f21c1874c98b5936e8a54b333b7 : SUCCESS in 23m 04s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/4848959bb0824855af8d28c348a292ca : SUCCESS in 33m 21s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/85b01f6dad5e43e9baf2292c4d59360e : SUCCESS in 27m 06s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/260684a23d3445cba99fd1ac38123874 : SUCCESS in 32m 24s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/639c7b43ac544d04b2b301346c29639a : SUCCESS in 28m 53s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/fb454415f6e24944bc6b791df53eb43c : SUCCESS in 31m 35s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/f4f0259a3833476eb1db37a87f07f2bd : SUCCESS in 26m 17s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/7f449d5106f44a9a862f159c6754aea8 : FAILURE in 15m 24s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/c9663d9d696f441a8e74342181ddda40 : SUCCESS in 18m 13s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/8bacf93110354d9f846272fe9aac7750 : SUCCESS in 23m 42s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/acc417ebd66c4074b76bc6b8ea9c9c1c : SUCCESS in 33m 26s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/7bfd954436984a97a75a1dd16c4f6e04 : SUCCESS in 15m 17s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/26f51fc3f63f44daafe720169df4b6e9 : SUCCESS in 18m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9c9e5f31862ce9802b06f7140855667f6631881b","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-09-26 16:51:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21b7d61b359eaaac9ac8c3cc6a778045c9d44a82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 17:30:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/376a8b6a86f8499fbd2a8e5a7c98f8d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37da36ebbf2448e580a58a2ee0e05d35 : SUCCESS in 2m 39s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/3a9482bb4e66453d8e3a1e3abdc44e53 : SUCCESS in 14m 56s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/07429eb18c83425b97f7497213885ba0 : SUCCESS in 29m 16s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/28ae6b02376f4e6d9c86104ea47bf0cd : SUCCESS in 14m 00s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/6bbc38de17a8424ea1f8b44ecc7a38e0 : SUCCESS in 15m 11s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/f48f331148ed44c6895bf2d80587bc31 : SUCCESS in 14m 26s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/71ca7cbfc4da400899c547d51b605a7d : SUCCESS in 30m 52s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/820205eb53e64cc988eb735a4ff83998 : SUCCESS in 15m 02s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/bce389f7fddc4d87a6f708054a4bb3dd : SUCCESS in 27m 26s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/503e2350246243da8f082a2a3ab1624f : SUCCESS in 14m 40s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/127b647ba29143bfaf2722a2d1bc8b70 : SUCCESS in 14m 30s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/3093f93e17b34316a80db12bf63d940c : SUCCESS in 14m 55s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/d7ced8dd42d0426193f52b7c7470815a : SUCCESS in 31m 14s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/55fbead980fb4c46ae4fb464bb8d66d5 : SUCCESS in 29m 20s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/3eebd903b734485b958c2c79d30b3890 : SUCCESS in 33m 54s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/994bf74c0f1b4874ac5c9f43be5c9fb8 : SUCCESS in 15m 37s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/bbfeea7bf8f54552b56ab4fa176ebd59 : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":2},{"id":"5c8ad24bf6a02751c6c96f09e9d69fd09878a5ce","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2025-10-07 18:38:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"89010184cfcc87613a085493412b5edb35166307","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 18:39:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"61b83a774eb5ba69320eb0d8bb9f9fb078856b9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:14:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12936b89bcdc4cc2be77e100306fdb0c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa54f43b1b344e9f9db7f70a67800328 : SUCCESS in 2m 47s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/7f2b63880fae4c91b319d6f39c44fb03 : SUCCESS in 20m 36s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/d735e260e4da4cb8a2b9f4bab5de6de9 : SUCCESS in 14m 05s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/4c85d14128ac44d7aeb60c4ed908e041 : SUCCESS in 14m 43s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/e643b89fe6e44ab18beaf17179a71b98 : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":2},{"id":"92020ad67580c15c40cb57d882f42087741457f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:14:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1f3368b08f4c17f9bcd8f11fb7f78fd5c173c89b","revisions":{"135946a72d96295475ebc9d4d2a2f6eb87907bdb":{"kind":"REWORK","_number":1,"created":"2025-09-24 21:25:32.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/31/962231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/31/962231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/31/962231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/31/962231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/31/962231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/31/962231/1"}}},"commit":{"parents":[{"commit":"b5c3246767cd4cf31d84aab9cfb3dee3520295bf","subject":"Add stable/2025.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/b5c3246767cd4cf31d84aab9cfb3dee3520295bf"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-09-24 21:24:34.000000000","tz":-420},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-09-24 21:24:34.000000000","tz":-420},"subject":"Remove api v1.1 support in tempest plugin","message":"Remove api v1.1 support in tempest plugin\n\nIt is for the task of removing old v1.1\nsupport in Zaqar project.\n\nChange-Id: I8f8045205b0fa1832118cd06fd1a1e0a9b81e96f\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/135946a72d96295475ebc9d4d2a2f6eb87907bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/135946a72d96295475ebc9d4d2a2f6eb87907bdb"}]},"branch":"refs/heads/master"},"1f3368b08f4c17f9bcd8f11fb7f78fd5c173c89b":{"kind":"REWORK","_number":2,"created":"2025-09-25 17:31:50.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/31/962231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/31/962231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/31/962231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/31/962231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/31/962231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/31/962231/2"}}},"commit":{"parents":[{"commit":"b5c3246767cd4cf31d84aab9cfb3dee3520295bf","subject":"Add stable/2025.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/b5c3246767cd4cf31d84aab9cfb3dee3520295bf"}]}],"author":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-09-24 21:24:34.000000000","tz":-420},"committer":{"name":"hwang","email":"sxmatch1986@gmail.com","date":"2025-09-25 17:31:37.000000000","tz":-420},"subject":"Remove api v1.1 support in tempest plugin","message":"Remove api v1.1 support in tempest plugin\n\nIt is for the task of removing old v1.1\nsupport in Zaqar project.\n\nChange-Id: I8f8045205b0fa1832118cd06fd1a1e0a9b81e96f\nSigned-off-by: Hao Wang \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/1f3368b08f4c17f9bcd8f11fb7f78fd5c173c89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/1f3368b08f4c17f9bcd8f11fb7f78fd5c173c89b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
