)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~962943","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I5c96e6abedd189f405d7ddf92816b732a50994bf","project":"openstack/zaqar-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-04 12:01:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I5c96e6abedd189f405d7ddf92816b732a50994bf","subject":"Remove useless cover target","status":"MERGED","created":"2025-10-03 14:45:49.000000000","updated":"2025-10-04 14:21:37.000000000","submitted":"2025-10-04 14:21:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962943","meta_rev_id":"4bee394f577b8336a31914bc8549e5aa15042bdb","_number":962943,"virtual_id_number":962943,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-04 14:21:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-10-04 13:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-10-04 13:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-03 15:43:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cdead6fc9aa730eec46f7b08fa0d4c0d0c3f399c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-03 14:45:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4837dfdc264d03a1b3449c91267e64804a2be76a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-03 14:49:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f05762584dd2ac877a90f7f904a7fbe5106eed67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 15:43:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d341a14b5cb42cba1b6e60141459843\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/00f10384cccb4850ab019be72874d483 : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cda143de81b54bc0b241849f262f250c : SUCCESS in 2m 42s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/21c8e4f2d18d4c18829f377a6fec7b7d : SUCCESS in 33m 33s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/f8ae4839c17741e3b65a9785567ea583 : FAILURE in 21m 34s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/01b092979ccc4badabc9879dd31afd30 : SUCCESS in 29m 15s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/5780c78abdf44802b78ce334579b8539 : SUCCESS in 24m 22s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/3cf2b51cc2e94350ae49cd9d36a2e789 : SUCCESS in 31m 12s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/8b7dc1a198684682ba0ecad76991e91b : SUCCESS in 31m 26s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/34dd6bb15c7e4d33a115747d4c7ab998 : SUCCESS in 40m 25s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/aeb4b0ed6a234eefbbce85557f5fc309 : SUCCESS in 27m 29s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/e0553b6b06544a83bc3a34a6ef194da7 : SUCCESS in 32m 36s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/928a3275143d4ec585bf2da927202ac8 : SUCCESS in 34m 25s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/78e1f45e843449588accb34cd44a802e : SUCCESS in 36m 12s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/934a5897964c48f1b97bfa4912ced580 : SUCCESS in 31m 25s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/c73b37b0d2714831a011fd3bae3bacbc : SUCCESS in 31m 56s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/d4479cb6c75f4a6c8d23510f3e2fbab0 : SUCCESS in 29m 49s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/e2212d8e2f8e460eb255258ff1c01877 : SUCCESS in 33m 37s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/3dea5771020a4ffca2546a742694d119 : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9582b92e16de874120cc711302a2b47c16243d3a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-03 22:04:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7e3d2a81239bc1c15236833d60207bae8b8caa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 22:45:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26e5a6f1cf344b6685df121efefbfd2a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a18568f9fbfa4922b32547d8799521d2 : SUCCESS in 1m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9261bfe59bd84533806cd50040c5aa78 : SUCCESS in 2m 33s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/23ca2d79352a41a1adc319223889d565 : SUCCESS in 23m 16s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/2c6045fae71b4f51b661ae66819a307f : SUCCESS in 32m 24s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/f6c6c13966ae43568ffefa823ea49f76 : SUCCESS in 22m 59s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/51acf184ae1247fc9854c618eda66851 : SUCCESS in 31m 55s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/51c53775555146029780b42fa1b18fa3 : SUCCESS in 25m 27s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/67394e95e55e46b681b3b731f7d2927f : SUCCESS in 22m 56s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/9ce54a08903d40c6ae5413bdc1745486 : SUCCESS in 33m 10s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/f220cc7d5dd94afda0ea680c0cb62d2c : SUCCESS in 20m 50s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/ee582dd7da694b2e8ff6f1474532fb0f : SUCCESS in 30m 29s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/1c4a0b3ad4ef4d20824692a8c5261d41 : SUCCESS in 22m 24s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/34c8c553328347e3a656d781ad39d71e : SUCCESS in 26m 43s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/c88a85751455409790c8da47d6c7c4a0 : SUCCESS in 22m 35s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/7305a359750445ab8ca96b9d284405ee : SUCCESS in 28m 58s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/5601fd735fd0495fb30856cf286d8834 : SUCCESS in 29m 06s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/826db956c2534b499af2e4f121fc25c9 : SUCCESS in 35m 59s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/20132390a459456aa9b949719a1cdf0b : FAILURE in 28m 31s","accounts_in_message":[],"_revision_number":2},{"id":"0664945d933ba1739a3491b079360a037347e636","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-04 12:01:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d8ce1e2c576512883be339595d23a622c497a8e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 12:44:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ca90b37a9634ae792a15f6c5da06939\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/469191f9c4374aaa8faedc9094efbc93 : SUCCESS in 1m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb8f0bec52904bdf9900bf627dd35e73 : SUCCESS in 2m 37s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/2f3265b913a947f7987ea0a9588eb74a : SUCCESS in 32m 09s\n- zaqar-tempest-plugin-mongodb-2024-1 https://zuul.opendev.org/t/openstack/build/aae84675220e48b29beeab61f1d6cdc6 : SUCCESS in 29m 19s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/0762d2102b68458dbfabc86106b2e9ca : SUCCESS in 27m 57s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/b9299ca35f414d888c22261f9e9381f8 : SUCCESS in 33m 45s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/06b0889b5c65408e9cb8791d3d7d5195 : SUCCESS in 24m 48s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/7cf78816493b4a1ea41615b4ee703451 : SUCCESS in 32m 27s\n- zaqar-tempest-plugin-redis-2024-1 https://zuul.opendev.org/t/openstack/build/61e46895ea864c8a8ad6fbbe0b39e2f5 : SUCCESS in 33m 54s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/70bb265e470e4dce8d5edb6109ef9814 : SUCCESS in 29m 52s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/c6079b44dde747fbb7e7fda708f213fa : SUCCESS in 35m 18s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/78135874a4a941b298cb80113ee4ea6d : SUCCESS in 19m 07s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/1e0aa6fcc0ef49d3b1967cea04c4b0a7 : SUCCESS in 26m 51s\n- zaqar-tempest-plugin-swift-2024-1 https://zuul.opendev.org/t/openstack/build/7b438c5a60b94d7db3c0c19098b32ca0 : SUCCESS in 32m 47s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/3e297191139e47e590fd9ea88c2fa333 : SUCCESS in 18m 58s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/3723b6c432944efead4fdf9ff38d0844 : SUCCESS in 21m 27s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/07e0ee321da2448687231bd707907b03 : SUCCESS in 35m 51s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/f9a8f485d7f34809ade54dcc6561d0f4 : SUCCESS in 21m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9e43162422c1f13f5ed77c9534690ac7a65968ef","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-04 13:43:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5cd2b8fc1ff061ed30f76fbd794522b8154d0b87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 13:44:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"58594a712e9f355bf8baee2ffcb89943c0fb5d5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 14:21:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8c3626d07b840f78876f2e3aaa6c798\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b883c39f11d40249e12fd53034ea8dd : SUCCESS in 1m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d0e2fa115984357b5db771f4d91ca6d : SUCCESS in 2m 25s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/c9d9e44a44654112acff46ca365a6afb : SUCCESS in 23m 26s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/ab8b463e1b3f4d88a90e6581443482e5 : SUCCESS in 30m 13s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/ffe961404552418cbd76f1bcd6cc7e5b : SUCCESS in 14m 52s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/f3633e004bec402abefea43b5aa8718b : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4bee394f577b8336a31914bc8549e5aa15042bdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 14:21:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0f1acb08af58f5e3854e706b915bea4e203bc773","revisions":{"230aaa1767a2a928ade54b26b3cc2651b6185872":{"kind":"REWORK","_number":1,"created":"2025-10-03 14:45:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/962943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/43/962943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/43/962943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/43/962943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/43/962943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/43/962943/1"}}},"commit":{"parents":[{"commit":"b5c3246767cd4cf31d84aab9cfb3dee3520295bf","subject":"Add stable/2025.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/b5c3246767cd4cf31d84aab9cfb3dee3520295bf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-03 14:45:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-03 14:45:37.000000000","tz":540},"subject":"Remove useless cover target","message":"Remove useless cover target\n\nThis repository contains code for testing and we have no real benefit\nwith code coverage.\n\nChange-Id: I5c96e6abedd189f405d7ddf92816b732a50994bf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/230aaa1767a2a928ade54b26b3cc2651b6185872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/230aaa1767a2a928ade54b26b3cc2651b6185872"}]},"branch":"refs/heads/master"},"0f1acb08af58f5e3854e706b915bea4e203bc773":{"kind":"REWORK","_number":2,"created":"2025-10-03 14:49:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/962943/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/43/962943/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/43/962943/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/43/962943/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/43/962943/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/43/962943/2"}}},"commit":{"parents":[{"commit":"b5c3246767cd4cf31d84aab9cfb3dee3520295bf","subject":"Add stable/2025.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/b5c3246767cd4cf31d84aab9cfb3dee3520295bf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-03 14:45:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-03 14:49:12.000000000","tz":540},"subject":"Remove useless cover target","message":"Remove useless cover target\n\nThis repository contains code for testing and we have no real benefit\nwith code coverage.\n\nChange-Id: I5c96e6abedd189f405d7ddf92816b732a50994bf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/0f1acb08af58f5e3854e706b915bea4e203bc773"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/0f1acb08af58f5e3854e706b915bea4e203bc773"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
