)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~967372","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~Ifc044dd6d5715623ef29f9102221fada52cc0fa3","project":"openstack/zaqar-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ifc044dd6d5715623ef29f9102221fada52cc0fa3","subject":"Skip functional tests for .pre-commit-config.yaml update","status":"MERGED","created":"2025-11-17 13:52:46.000000000","updated":"2025-11-17 15:59:12.000000000","submitted":"2025-11-17 15:59:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967372","meta_rev_id":"a9e1836fe0ec1029d1cc799a135e5fb726f681cc","_number":967372,"virtual_id_number":967372,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 15:59:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-11-17 15:18:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-11-17 15:18:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-17 14:45:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f8b2e81aa18a467be3c29ad48cd280652f5d589","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-17 13:52:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"540476f93bfbd6cba2a333acaa1c6d48db298893","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 14:45:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36798fa52f314684b877f39ddd61ff30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8059ec7eb4457c9f51c632d2a8d9e6 : SUCCESS in 2m 09s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/3ed25f3a2f264ed6ae5c897d7cd6b12d : SUCCESS in 15m 19s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/96d43212012e4db1a743db313af9ef6b : SUCCESS in 14m 28s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/ada719fc6f254d5c89efbb764a56467a : SUCCESS in 32m 45s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/c3f51163f58145129428577ad5eb5e97 : SUCCESS in 33m 13s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/f3d61a9c7e8b427a933669f86c7704e5 : SUCCESS in 31m 25s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/95fb9310e57f4a91b4368a0549712aaf : SUCCESS in 22m 10s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/996a3c2bfebd48ef9d6510f00f415c40 : SUCCESS in 15m 21s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/2caae8e439c5443db2fe3117831325c9 : SUCCESS in 34m 45s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/5d295369185345b58f1e96ece445517a : SUCCESS in 23m 46s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/8282e963cea14eaa90f8b6c415dcc20f : SUCCESS in 34m 42s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/09877a5530534845aa8a706dd3c3c815 : SUCCESS in 39m 28s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/b240278bfef74949af259d3780b891f9 : SUCCESS in 32m 21s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/abaa479a429747e280b91963449f67b1 : SUCCESS in 27m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7c9eecbe509a2e32c3cc45be002a1bec399fb991","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-17 15:18:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"854f894d77b9e15d7132c37a9bd296778f007e21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 15:19:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff38299e46bb00d5d07e0673d03b21708cf4f5cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 15:59:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6897401f3717498cb4067b715980d538\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da10fdefb1ae4b469f78b0daa6e2efa9 : SUCCESS in 2m 43s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/c95c9754adeb4677b409e2ae5c7b27e9 : SUCCESS in 33m 19s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/f44048832d5f42d0b6a9c9da4437a4ce : SUCCESS in 28m 41s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/559cf35359534f71933046efba924383 : SUCCESS in 16m 23s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/96bc73d27ac54b0d851c638e1063a1ee : SUCCESS in 16m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a9e1836fe0ec1029d1cc799a135e5fb726f681cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 15:59:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a52c5fe7b52c60afcd1df9fc8eea89f8b090d683","revisions":{"a52c5fe7b52c60afcd1df9fc8eea89f8b090d683":{"kind":"REWORK","_number":1,"created":"2025-11-17 13:52:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/967372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/72/967372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/72/967372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/72/967372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/72/967372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/72/967372/1"}}},"commit":{"parents":[{"commit":"691db5e68e5bbae56383e65af94787eddd1571fa","subject":"Update stable branch jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/691db5e68e5bbae56383e65af94787eddd1571fa"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-17 13:52:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-17 13:52:37.000000000","tz":540},"subject":"Skip functional tests for .pre-commit-config.yaml update","message":"Skip functional tests for .pre-commit-config.yaml update\n\n... because the file is used only in pep8 jobs.\n\nChange-Id: Ifc044dd6d5715623ef29f9102221fada52cc0fa3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/a52c5fe7b52c60afcd1df9fc8eea89f8b090d683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/a52c5fe7b52c60afcd1df9fc8eea89f8b090d683"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
