)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~980978","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I611b671b4fae13e6a5f6c105c42ef90ed43b3da5","project":"openstack/zaqar-tempest-plugin","branch":"master","topic":"2026.1-job","hashtags":[],"change_id":"I611b671b4fae13e6a5f6c105c42ef90ed43b3da5","subject":"Add stable/2026.1 job","status":"MERGED","created":"2026-03-17 15:23:17.000000000","updated":"2026-03-17 17:43:33.000000000","submitted":"2026-03-17 17:43:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980978-2026.1-job","meta_rev_id":"3d60cf3a55c887aa0c65ff037438b061433577f8","_number":980978,"virtual_id_number":980978,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-17 17:43:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-03-17 17:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-03-17 17:03:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-17 16:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31d32bad13adc66e6bfa36bf8bbc6f2cf306e7cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 15:23:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0140308a4014e5fc2fa39d6b600e151274a077b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 16:11:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5ddd4223e624a959fef64f05f5911d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793cca9e7bbc40d89c0ba97faa981589 : SUCCESS in 2m 51s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/8f16040dd1854364bd0831ea0894a87e : SUCCESS in 35m 06s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/8606f18ea76248d6a08b0ecf965dbfbe : SUCCESS in 30m 34s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/3ec4893fe24a4dcdb1c0b8ac25d1e8e4 : SUCCESS in 36m 07s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/543e70b9b17443619009da6891f7c4be : SUCCESS in 21m 49s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/8d2cb1794567452ca9823327d8968dba : SUCCESS in 26m 00s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/a29d4d02a83e48849250f92998664b86 : SUCCESS in 25m 40s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/426a4576f9d14d2eb50aea6c2f2a759b : SUCCESS in 31m 50s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/4db99c5741bc4fd0a0ff2db70811e88f : SUCCESS in 37m 23s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/5439833032bf486caa7efda19bc69138 : SUCCESS in 31m 55s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/91d4442ea0764969979aaf3346a1e013 : SUCCESS in 22m 18s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/d07bdd910e164d7d94e2d49f056d2555 : SUCCESS in 21m 52s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/e4651620248946a1967f60863add8870 : SUCCESS in 20m 31s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/8d29eac491c846deb71cac236bdf588a : SUCCESS in 36m 30s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/2a6b3cf4601d4595a4f938c78f66a950 : SUCCESS in 22m 23s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/deaf40cac8fb4e8e958ae3b8bb488dd7 : SUCCESS in 38m 26s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/a9d9c95e34f3471a8865ad22bf1908ad : SUCCESS in 27m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e186d4c0ed624b69b57afd99199f973fb9ff8b28","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 17:03:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83b6fa50e1b929ca2769d9ca0a51193f3b0f45be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 17:04:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb884bf1453bbc7bbfd575898313d804a5b5a6cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 17:43:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf63784569d04411a3c5a63b9e37a117\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d293d04d9d5f43eb952626f619b0ad84 : SUCCESS in 1m 41s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/85fbcca130354de0a3dda5da5f475b9e : SUCCESS in 34m 22s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/06366b2b5f92421ea8c38e329ee4038a : SUCCESS in 15m 28s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/b8dea6d59d084a04abbd8005448c7cde : SUCCESS in 16m 24s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/8b384109f6cb4049b0d7b501fbfe04c2 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3d60cf3a55c887aa0c65ff037438b061433577f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 17:43:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8a3718f96d29c2723d8ba5ceb8e80e607df38ce","revisions":{"a8a3718f96d29c2723d8ba5ceb8e80e607df38ce":{"kind":"REWORK","_number":1,"created":"2026-03-17 15:23:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/980978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/78/980978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/78/980978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/78/980978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/78/980978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/78/980978/1"}}},"commit":{"parents":[{"commit":"7a45cac86afe6f32fdbf2b0dfb81da988f7a2dff","subject":"Skip tempest jobs for .gitignore update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/7a45cac86afe6f32fdbf2b0dfb81da988f7a2dff"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 15:23:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 15:23:08.000000000","tz":540},"subject":"Add stable/2026.1 job","message":"Add stable/2026.1 job\n\nThe branch was cut, so should be tested.\n\nChange-Id: I611b671b4fae13e6a5f6c105c42ef90ed43b3da5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/a8a3718f96d29c2723d8ba5ceb8e80e607df38ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/a8a3718f96d29c2723d8ba5ceb8e80e607df38ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
