)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~987024","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~I8c0e5b9176df055e450f16d922e9ce9433bcbb5d","project":"openstack/zaqar-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-03 05:25:42.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I8c0e5b9176df055e450f16d922e9ce9433bcbb5d","subject":"Disable cinder","status":"MERGED","created":"2026-05-03 02:37:41.000000000","updated":"2026-05-03 07:22:34.000000000","submitted":"2026-05-03 07:22:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987024","meta_rev_id":"653b2e6c1ea689e6e32e8c9c4efec102e5168d67","_number":987024,"virtual_id_number":987024,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-03 07:22:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-05-03 06:42:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-05-03 06:42:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-03 03:33:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac00534138a937aafb92aba4c8f3c2a21177746c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-03 02:37:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4da7e06a74047cfdbf3b2ad7563394dfbab3417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 03:33:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83de08d793c044b687e18cc8631f8c94\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/740eed663cdc4d389b79344970a600d7 : SUCCESS in 4m 08s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/89c929ffc38045f09a527cba4eca297d : SUCCESS in 32m 26s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/5c9b95c803f6459ab305c04b878cfba7 : SUCCESS in 30m 38s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/f6372eed3b7b48e08d4790b0571715b8 : SUCCESS in 30m 40s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/be5732f7cb6f40369f48b86ff02a27fb : SUCCESS in 30m 17s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/2cbb0965dc2a452c8afa23b5b18e90eb : SUCCESS in 31m 37s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/5ea02d968a894d80ba8655ad8fcadcf0 : FAILURE in 31m 26s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/ff47740e4f57461d97a66afa1972fb93 : SUCCESS in 16m 04s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/2b9f7a415eb4400e8fb73159262354b2 : SUCCESS in 30m 48s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/736dbfa7f295404e900fa5ba18aabafe : SUCCESS in 18m 26s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/1c5ea8f9e4b64fbc9e3b85b538e6f739 : SUCCESS in 30m 25s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/f95bbee98f0f49408aecce8918b7fcfc : SUCCESS in 34m 31s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/d3a673438e954d5682095f22f5cee1d5 : SUCCESS in 15m 44s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/3c5196fc5dd9471abf44586fea2c2ad7 : SUCCESS in 18m 20s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/2354d619467045458b60c1e26963dfe2 : SUCCESS in 34m 03s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/f26d614fb04647879eade56bba9d95b7 : SUCCESS in 25m 09s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/3d693ae1326944f68bb31bf30d0c5474 : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c159548cde6ed913d4959f1ad3863e5801774882","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-03 04:42:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"761db702e77f4973e5b2ef8c3bb69db49a9f34e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 05:24:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ca6afc5a81f4cb9b5f3a9c31bf35ee4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7609cf6aec407fbeecf5815e2a04ef : SUCCESS in 2m 20s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/1954bdb465b64050ae32b1b673886035 : SUCCESS in 19m 14s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/738c79b7e84f474d9580224c6a647767 : SUCCESS in 29m 08s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/bd5a0235c1514d82b826f26ad953e9c1 : SUCCESS in 31m 59s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/9cd764246d034f21a3ab689a3f465c7f : SUCCESS in 15m 09s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/e50479dfe85040d49b59b35c4ce3802a : SUCCESS in 31m 36s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/087f5e1036d74d5ca843c78c0b274499 : SUCCESS in 18m 47s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/ff241042e9754a2ea575e280fd592e06 : SUCCESS in 27m 02s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/90c958f2e1794920acccb1d0155ffc6d : SUCCESS in 14m 37s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/4d12d9f79f164ecb952220b1dbf64ba6 : SUCCESS in 30m 18s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/b881607cae2f4ee5a6c1aebea7721e14 : SUCCESS in 14m 58s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/707e9b646bf34cfdad1906b1241ec29b : SUCCESS in 19m 52s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/e21cf66e56404e7e9533073f83071a75 : SUCCESS in 29m 52s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/c460d8c882064313871ee736e95e8797 : SUCCESS in 32m 06s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/d1d478483b33437f9c82c0a608153b64 : SUCCESS in 19m 31s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/a931808b4774442ba233177dc192e4c0 : FAILURE in 36m 29s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/b123cc9ceeba4f0997fd07ab1001d649 : SUCCESS in 19m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9f0d04b73edd6ad039b7402e562efe27fa9256c5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-03 05:25:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5c2b90bcfb56cb088a948b032ca0f0386fedcec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 06:08:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0d337ff64be4ea0a8a4a224ba807bc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c5718f50ef248519d6cd9f16bbde0d2 : SUCCESS in 2m 38s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/86cf336dd0224b5083abdb66a0b63d96 : SUCCESS in 34m 18s\n- zaqar-tempest-plugin-mongodb-2024-2 https://zuul.opendev.org/t/openstack/build/eff956de246048768930768b7f93b10e : SUCCESS in 19m 24s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/5bd5eddd1bf342efafabfa345137b5b0 : SUCCESS in 32m 23s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/4e8bd43d2e7148619d421e77cfd20bbd : SUCCESS in 19m 59s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/8cbe3123278c4069bf23b97229978da4 : SUCCESS in 33m 53s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/4ec21255af314bb48de3db1b02ee3ac4 : SUCCESS in 32m 39s\n- zaqar-tempest-plugin-redis-2024-2 https://zuul.opendev.org/t/openstack/build/fb5483d312724af29073b3c30f9aa2e3 : SUCCESS in 28m 54s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/4fed060c48e14cf1aac2d14e73e0646e : SUCCESS in 27m 35s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/b09cbaa2c33740638cd1e7cd0736f5ba : SUCCESS in 29m 31s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/c8e550e9192b4eda979ed4f404cbcc6e : SUCCESS in 27m 13s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/9a7df9cc358e4ba185fb550f71b18e4a : SUCCESS in 31m 41s\n- zaqar-tempest-plugin-swift-2024-2 https://zuul.opendev.org/t/openstack/build/2d91f1d6a88841e39ddf84392fda2656 : SUCCESS in 28m 49s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/2af9c92f6e19474796392bd8c85a3f81 : SUCCESS in 33m 02s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/24e2f9b45b064906821292e19b4965cb : SUCCESS in 28m 56s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/cb78256898dd431aac822f0fc7677429 : SUCCESS in 37m 05s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/f049e2a040834708abd33860713f6f3a : SUCCESS in 33m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f3d0ed8fa62edf3c6eb4860458470c017bad10af","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-03 06:42:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cfd89c5564f0bb19bd28981fd74541a7c4866d59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 06:43:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e9a7d101bf40438c78bd14613c65ca1c257d883","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 07:22:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72d27682d7734bc6ac5ad04c4e93e137\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95dc9fb4bac14f6fb50c124342f17043 : SUCCESS in 2m 31s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/5ec94e0e49634d83bcce5521f5b1f7f9 : SUCCESS in 22m 22s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/0fc30a04e771475e99b6d5a18aac9de6 : SUCCESS in 17m 22s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/efbff6532e214dec84261855bced6a11 : SUCCESS in 34m 38s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/3ca2eafd2d0845b593451166420f1b33 : SUCCESS in 23m 30s","accounts_in_message":[],"_revision_number":1},{"id":"653b2e6c1ea689e6e32e8c9c4efec102e5168d67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-03 07:22:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0cf5b4ef358e2d096a2a299897a2ff30ca286198","revisions":{"0cf5b4ef358e2d096a2a299897a2ff30ca286198":{"kind":"REWORK","_number":1,"created":"2026-05-03 02:37:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/987024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/24/987024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/24/987024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/24/987024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/24/987024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/24/987024/1"}}},"commit":{"parents":[{"commit":"a8a3718f96d29c2723d8ba5ceb8e80e607df38ce","subject":"Add stable/2026.1 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/a8a3718f96d29c2723d8ba5ceb8e80e607df38ce"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-03 02:37:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-03 02:37:07.000000000","tz":540},"subject":"Disable cinder","message":"Disable cinder\n\n... because the service is completely irrelevant.\n\nChange-Id: I8c0e5b9176df055e450f16d922e9ce9433bcbb5d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/0cf5b4ef358e2d096a2a299897a2ff30ca286198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/0cf5b4ef358e2d096a2a299897a2ff30ca286198"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
