)]}'
{"id":"openstack%2Fzaqar-tempest-plugin~987444","triplet_id":"openstack%2Fzaqar-tempest-plugin~master~If5e9434ab1e22029b181b3caa7188e8564137bdf","project":"openstack/zaqar-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-06 12:18:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"If5e9434ab1e22029b181b3caa7188e8564137bdf","subject":"Refactor register_opts","status":"MERGED","created":"2026-05-06 03:40:14.000000000","updated":"2026-05-06 13:01:18.000000000","submitted":"2026-05-06 13:01:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987444","meta_rev_id":"84529026be8676e82201de138918b7e4ce1004cb","_number":987444,"virtual_id_number":987444,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-06 13:01:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-05-06 12:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-05-06 12:18:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-06 04:10:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a9043f937b0521b68e220bb2b9dfdf083cfb8ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-06 03:40:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76b3ffce6dc5e7eeff6c4efdd71a7abeb84fd624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 04:10:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37b98e7724d84e12a7ed07a4f46b0ee6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f39f1041ce08447ba51e552c882d00f0 : SUCCESS in 1m 39s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/a8afa62bc3f94c74b0da94818c668bdc : FAILURE in 16m 39s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/6b37eb7acc5c4a9abfacf7d4e680904d : FAILURE in 16m 47s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/cdba036ea1a443ff869afb6299a1c623 : FAILURE in 18m 30s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/7c251b4b66e24de1a5139cd49e460d62 : FAILURE in 23m 07s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/8c52764e28ce4603bfa4e67442789193 : FAILURE in 17m 28s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/9bf243c8cc0e41ab89171e8255236835 : FAILURE in 16m 43s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/9002092e566e4906b709ef23b7557276 : FAILURE in 16m 23s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/701ea82900ba423abba5a34a30d13b26 : FAILURE in 26m 47s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/e612035f67494a309dd48d9cc32d95e9 : FAILURE in 12m 35s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/8fbc17c647244314afea09b9a2c323e6 : FAILURE in 28m 41s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/972d66bfae10406a8b3cdaeff002533a : FAILURE in 11m 54s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/18e43f2762224c5087ef0bc31f64ab42 : FAILURE in 17m 21s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/2255bc8f0485458fb0c2e65c10de80b4 : FAILURE in 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3be06835331d8173bd8ae48ab82707b4346f2303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-06 04:34:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fd4a01289a447087ccdbb147a345c0d87ca949df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 05:14:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a489601ab2cd45d89b59680fd68c7bb7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f4d9691371741388a8ca6f73d0463b4 : SUCCESS in 3m 37s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/f4834fede62b464585d5bc7a1a6c4efb : SUCCESS in 32m 43s\n- zaqar-tempest-plugin-mongodb-2025-1 https://zuul.opendev.org/t/openstack/build/ca6adf6841cf435b97c74d3ae90ed1c4 : SUCCESS in 29m 35s\n- zaqar-tempest-plugin-mongodb-2025-2 https://zuul.opendev.org/t/openstack/build/fd295ce95b2d4d64a99ea62fa139216b : SUCCESS in 18m 03s\n- zaqar-tempest-plugin-mongodb-2026-1 https://zuul.opendev.org/t/openstack/build/4048ee74e07c46c99b18759edf823b2c : SUCCESS in 25m 08s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/714e30e23a69420a9fd54473a1151f7d : SUCCESS in 29m 57s\n- zaqar-tempest-plugin-redis-2025-1 https://zuul.opendev.org/t/openstack/build/2349376d3e8e48d289fe412d7775d6e2 : SUCCESS in 29m 17s\n- zaqar-tempest-plugin-redis-2025-2 https://zuul.opendev.org/t/openstack/build/a057d26a64264a8c96550133277f7e91 : SUCCESS in 15m 24s\n- zaqar-tempest-plugin-redis-2026-1 https://zuul.opendev.org/t/openstack/build/d10c7c49ec9b4e8786c098c4a6f08b98 : SUCCESS in 29m 07s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/3bd9203c84ab4c45b3891854cebf2afa : SUCCESS in 34m 12s\n- zaqar-tempest-plugin-swift-2025-1 https://zuul.opendev.org/t/openstack/build/52b69f39e28e468bbdeea8b65d81fcb8 : SUCCESS in 31m 10s\n- zaqar-tempest-plugin-swift-2025-2 https://zuul.opendev.org/t/openstack/build/5b3d2ddaf22b4f62a348ba0531e78f60 : SUCCESS in 20m 03s\n- zaqar-tempest-plugin-swift-2026-1 https://zuul.opendev.org/t/openstack/build/2666bfff4c274674963ea8ece6bc21c5 : SUCCESS in 32m 28s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/236c5615663747a7a01f2194ef6aee78 : SUCCESS in 32m 12s","accounts_in_message":[],"_revision_number":2},{"id":"30acafebdf97b14f4fa5b26e2db8d31d36a53669","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-06 12:18:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d632744132bba7c9c1b9376206a3ac43d4861bb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 12:19:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bbab6a88edfd2c5df409d58f167d4e98860e5e72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 13:01:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31429a1d86ac40559c646152eb997c9b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff3630878fe4317b8d14872969bae53 : SUCCESS in 3m 03s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/43dd86b5aad549d68210124e8cdac1a2 : SUCCESS in 34m 26s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/63000824f6c6432aad31611b11f2848e : SUCCESS in 30m 53s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/e5b95e9e25154b61ac5b56ceff6d0616 : SUCCESS in 20m 59s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/68c1511c620640dc9da4152fa092999b : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":2},{"id":"84529026be8676e82201de138918b7e4ce1004cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 13:01:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8fcddaa3f59ae565b2fa1cb8f430a9626c9bf35f","revisions":{"4c9e36b47d37863f93b6091f7e8b1286603a09c8":{"kind":"REWORK","_number":1,"created":"2026-05-06 03:40:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/987444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/44/987444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/44/987444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/44/987444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/44/987444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/44/987444/1"}}},"commit":{"parents":[{"commit":"d585f68ae08d77c15751f8fd92d9da8153f39293","subject":"Merge \"Disable cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/d585f68ae08d77c15751f8fd92d9da8153f39293"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-06 03:39:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-06 03:40:04.000000000","tz":540},"subject":"Refactor register_opts","message":"Refactor register_opts\n\n... using the core tempest utilities.\n\nAlso add documentation of the plugin class for easy reference.\n\nChange-Id: If5e9434ab1e22029b181b3caa7188e8564137bdf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/4c9e36b47d37863f93b6091f7e8b1286603a09c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/4c9e36b47d37863f93b6091f7e8b1286603a09c8"}]},"branch":"refs/heads/master"},"8fcddaa3f59ae565b2fa1cb8f430a9626c9bf35f":{"kind":"REWORK","_number":2,"created":"2026-05-06 04:34:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/987444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-tempest-plugin","ref":"refs/changes/44/987444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/44/987444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/44/987444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/44/987444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-tempest-plugin refs/changes/44/987444/2"}}},"commit":{"parents":[{"commit":"d585f68ae08d77c15751f8fd92d9da8153f39293","subject":"Merge \"Disable cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/d585f68ae08d77c15751f8fd92d9da8153f39293"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-06 03:39:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-06 04:34:42.000000000","tz":540},"subject":"Refactor register_opts","message":"Refactor register_opts\n\n... using the core tempest utilities.\n\nAlso add documentation of the plugin class for easy reference.\n\nChange-Id: If5e9434ab1e22029b181b3caa7188e8564137bdf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/8fcddaa3f59ae565b2fa1cb8f430a9626c9bf35f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-tempest-plugin/commit/8fcddaa3f59ae565b2fa1cb8f430a9626c9bf35f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
