)]}'
{"id":"openstack%2Fzaqar-ui~692012","triplet_id":"openstack%2Fzaqar-ui~master~Ia3f663e1271deb7a7aef6ee5fa1b17569fa5ce0e","project":"openstack/zaqar-ui","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Ia3f663e1271deb7a7aef6ee5fa1b17569fa5ce0e","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2019-10-30 03:12:03.000000000","updated":"2019-11-26 04:41:24.000000000","submitted":"2019-11-26 04:39:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692012-1574743143347-12b15d9b","meta_rev_id":"ba97611fe10f63ad55c06d727f068400ad5e1886","_number":692012,"virtual_id_number":692012,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 04:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":2,"date":"2019-11-26 04:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":1,"date":"2019-11-21 15:16:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2019-11-22 15:17:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":1,"date":"2019-11-26 04:29:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 15:16:15.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2019-11-22 15:17:33.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-11-26 04:29:03.000000000","updated_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"reviewer":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"state":"REVIEWER"},{"updated":"2019-11-26 04:39:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01ae59b165143cad88646d58c3ae447438b4f153","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-30 03:12:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f69aedb59a567faa463a93e742eff44718f0a7e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-30 03:24:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fda6d66aef0c44a17098d12428b161a3275fbebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 03:34:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c9d232d0cd5c4c5da09ac97bb0436bea : SUCCESS in 2m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92cbf1cf8b7a4faf95bd1ea06a39df71 : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/321630c1abeb49f9b93679631330358b : FAILURE in 2m 49s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/73225cc6a35e4ef1a43ec41fcdf5f55b : SUCCESS in 4m 10s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/dff0f99e9a864236b5123b7dc0f9a011 : SUCCESS in 5m 10s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/987bd638c5c64990a77fb407bbbc8289 : SUCCESS in 5m 41s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/b0f2112e31ee4050b843a3fc9dbf09dc : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14585ac600bd4c2aa0ddd1585b357bc2 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e25798a9e33f44ebb17cf64d9833efa1 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/597ba039e0384405a7f4be4d3b7fab1c : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4b745d3de5b42f18dba2033de351ef3 : FAILURE in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75321ae61e31424c931b9a9258dd4a7c : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1f036d103bdd6496174b008397808005d36aec87","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-30 06:12:28.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3211db345d6c03e41905d2fd82fcd4443611ee1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 07:40:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0818500ea1b044c690c8550a7385a050 : SUCCESS in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db032408cf434e2caf98f423ef162e33 : SUCCESS in 4m 18s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/f1ad3db140444977956bed127a82c594 : SUCCESS in 4m 39s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/6414b5de21124a6d94a8c717070c446d : SUCCESS in 4m 24s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/ef4c6edf92b341a4b4d4a774c73d913a : SUCCESS in 6m 24s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/7da69fb79ec64ca1a4675d6ecea841ad : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f30c00089a4959baaa54b22f5513de : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a48c1bed8f63458cb1969ebc38d4b072 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/257e2c31251940ad8bf3896fb234a510 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b989c0ebf224d26bf487d367e45182a : FAILURE in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ccecdac07db46e8a93001a4ba5d19a9 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"c7c3d7b8c760a3e2bcb28e379bb758d094fcb33a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 00:50:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a248330eeda6ac4927ea1dc546157f1f20ef0b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 01:01:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b723ccb04e584e4a95cc345e0e143446 : SUCCESS in 1m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6ba31ac93f74689998f09210229783f : SUCCESS in 4m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3bb7238edf5248a5bcf745a5b67f8bf9 : SUCCESS in 1m 45s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/4336c176a3ca4b358ccc846ea1792fe7 : SUCCESS in 3m 41s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/cb6859cecdc740d39bc4c2d14175fbb5 : SUCCESS in 3m 46s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/e9087dd2f27a49759d91fdddcc2a8d42 : SUCCESS in 4m 46s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/c0364dfed6344d888fd416509fa395ef : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b1b59d8d5ec4e4294bdb09fa32396cf : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0d68a8ea6934245afe9f2d3d8a6c9bb : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd34dd10ff484fc58825384152842f2d : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53fd86e85d345a1becea3795d2d3b5f : FAILURE in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/628dbcefd3e2487d9f6bbab5c3acfd1f : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":4},{"id":"cfaacc8c01981cb056962d7c578de5bc90dbdb95","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-16 17:34:44.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a843e2ef9469576b9cbd76825ab4562452490cc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 17:43:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/58edaac6e1d44c1aa86ec904bc532be3 : SUCCESS in 1m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8be90ca4edc44d382d503b225d73091 : SUCCESS in 5m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7721cd6011fb48dc99a6ae03bacaed75 : SUCCESS in 1m 46s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/0c07085382c3448c9fe3c07d72b6b363 : SUCCESS in 3m 58s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/4e079e9fce014e25a44e314b20a3c0c5 : SUCCESS in 4m 00s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/34ed15621a904046a262fd5fa1e4bcc5 : SUCCESS in 4m 47s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/ea5d2c66f8c842ff87a6c8db0d358773 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2442bd2b0041288f20155fee592d80 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68566d7f730248bca1807e37ea18cff2 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62f59e88cd484366a13dfa327812801c : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92694d86f744432a8bdc119b8ca4c28c : FAILURE in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/879ebdc74b0d4d2ea2af5a14ce56a17e : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":5},{"id":"85fb56594a5e2cd90d5d0bacb583f301c2e87b4a","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-21 10:22:12.000000000","message":"Patch Set 5:\n\nrecheck.","accounts_in_message":[],"_revision_number":5},{"id":"c3eda1bb7e5c52c05384abe8f0f717280c4f66db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 10:32:35.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1f2dd90c229c425aa40e57cfc6819a5f : SUCCESS in 2m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36a28b72dc3f44bd890279ad034e7e2a : SUCCESS in 4m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/93b3790ca74d450c9595eb4605f5f42e : SUCCESS in 3m 07s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/58ef829330954a3391a3d642b576c5ba : SUCCESS in 4m 07s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/a6e2a92ce231403cb358cdf52446ca7d : SUCCESS in 5m 26s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/41f550af4ea84830a8c67bc5f9ab9225 : SUCCESS in 5m 14s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/1839bdc34ec14d91b2cec8aed82b0232 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6b94a0282c74a60a045cbc0b08b01b3 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb75e13262c342dd970a1ffd932c44df : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/969ed2ed7b3e405aa6b06105b9a5efc5 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25c1e5f9b4c441a9524a4e78728e0ac : FAILURE in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93951d23db76423ab1dd7627a4b76628 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":5},{"id":"9e4244a6437c6e5f8e7208860fd63150f01a5e5f","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-21 11:37:05.000000000","message":"Patch Set 5:\n\nrecheck.","accounts_in_message":[],"_revision_number":5},{"id":"3a3061e16c8b2e671256ba745a9ba60bb6db68a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 11:49:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9dc52da57ae64c3db7064e6fd3b872fb : SUCCESS in 2m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e114bf7405bc48258046ff98beebc2c7 : SUCCESS in 4m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/06e9e54ad8d94975a0f6ce7b2a2d7dca : SUCCESS in 1m 41s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/e24f135dd7874178a82dccbb775496db : SUCCESS in 4m 52s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/ede819fca38b4a41aadd89f73acfdbb3 : SUCCESS in 4m 05s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/ffcf14f93aad4d02a877dec5ae364062 : SUCCESS in 4m 56s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/5a6db399ae76483c98a7fb41fc20b628 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca00889408094c37a144fd98be530a67 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/434b08a82cd84e2b88b07e73ad7c8a37 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba46430774b44216a92cb708e4c3266f : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8a50d2518a4c1c948b73a745638420 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7252ed1eb3c4dcd8066175e99a8e476 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":5},{"id":"405eeec25fffc66a36539cf4a3ac7b4cbcfe3717","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-21 15:16:15.000000000","message":"Patch Set 5: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":5},{"id":"5487bdc4e11e2c1de0dbad8a1ee1bc0c3c69287f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-11-22 15:17:33.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"03c43c87d2c2a95cadb0978c30bd80670486ceb5","author":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"date":"2019-11-26 04:29:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a4f8c59510b803523ada14d8232611649f7f8e97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 04:29:18.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d121ece880ddef56c8f4f97beadf77b49a356527","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 04:39:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"9ab034db69284eb4ff089e8dffa9d37c41f2f36b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 04:39:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ddcc997072054a348b9cba14067cce6e : SUCCESS in 2m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63417c9177ce438cabd31c968552ef8b : SUCCESS in 3m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5fdbee15b71c49ed9262eb141af2aaba : SUCCESS in 1m 54s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/c68f1f02b32242798fcf37045419b48e : SUCCESS in 4m 14s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/1f16bd2fc2214a0ab500fb3ed20b6a58 : SUCCESS in 3m 50s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/2c73d3717e5b403393338f9a328f0c7d : SUCCESS in 5m 02s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/c2edc4fa16a44e0380ac05391b14c163 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c8ca716386646548b11252361712fa1 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce2d3e0e0cb94a4897bb776393352f49 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/269ac9c11d544af289630050d346986a : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c111185a0d574c1db3fcb015f4a98236 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2ad1cf9c0e8475483a6f66d2d78c353 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":5},{"id":"8ccc49ad0805e3c796f41c27d1648b40f256be71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 04:41:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a257615770048278453bab4c5a2ac53 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d613a5ae1a3440f29b53e158997bf8cc : SUCCESS in 59s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f8a51b3e4ff4a0c851d23a8c247e38ac612e3718","revisions":{"0da056076c9c17bd02d3b22d1037696ff4b62150":{"kind":"REWORK","_number":1,"created":"2019-10-30 03:12:03.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/692012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-ui","ref":"refs/changes/12/692012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/1"}}},"commit":{"parents":[{"commit":"835eac405b9777e2ff20802b99c187de716995e6","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/835eac405b9777e2ff20802b99c187de716995e6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-30 03:11:18.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-30 03:12:01.000000000","tz":0},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nzaqar-ui is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: Ia3f663e1271deb7a7aef6ee5fa1b17569fa5ce0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/0da056076c9c17bd02d3b22d1037696ff4b62150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/0da056076c9c17bd02d3b22d1037696ff4b62150"}]},"branch":"refs/heads/master"},"c81258b13726aa227785adab8bb6dc288cba4d38":{"kind":"REWORK","_number":2,"created":"2019-10-30 03:24:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/692012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-ui","ref":"refs/changes/12/692012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/2"}}},"commit":{"parents":[{"commit":"835eac405b9777e2ff20802b99c187de716995e6","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/835eac405b9777e2ff20802b99c187de716995e6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-30 03:11:18.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-30 03:24:32.000000000","tz":0},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nzaqar-ui is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: Ia3f663e1271deb7a7aef6ee5fa1b17569fa5ce0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/c81258b13726aa227785adab8bb6dc288cba4d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/c81258b13726aa227785adab8bb6dc288cba4d38"}]},"branch":"refs/heads/master"},"fdd3f6d48d8b64403c495656c0fa921bde6ca395":{"kind":"REWORK","_number":3,"created":"2019-10-30 06:12:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/692012/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-ui","ref":"refs/changes/12/692012/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/3"}}},"commit":{"parents":[{"commit":"835eac405b9777e2ff20802b99c187de716995e6","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/835eac405b9777e2ff20802b99c187de716995e6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-30 03:11:18.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-30 06:12:25.000000000","tz":0},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nzaqar-ui is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: Ia3f663e1271deb7a7aef6ee5fa1b17569fa5ce0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/fdd3f6d48d8b64403c495656c0fa921bde6ca395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/fdd3f6d48d8b64403c495656c0fa921bde6ca395"}]},"branch":"refs/heads/master"},"a9057604c1dab174492e94ee7c5821a602651631":{"kind":"REWORK","_number":4,"created":"2019-11-15 00:50:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/692012/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-ui","ref":"refs/changes/12/692012/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/4"}}},"commit":{"parents":[{"commit":"835eac405b9777e2ff20802b99c187de716995e6","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/835eac405b9777e2ff20802b99c187de716995e6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-30 03:11:18.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-15 00:49:13.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nzaqar-ui is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\n\nChange-Id: Ia3f663e1271deb7a7aef6ee5fa1b17569fa5ce0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/a9057604c1dab174492e94ee7c5821a602651631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/a9057604c1dab174492e94ee7c5821a602651631"}]},"branch":"refs/heads/master"},"f8a51b3e4ff4a0c851d23a8c247e38ac612e3718":{"kind":"REWORK","_number":5,"created":"2019-11-16 17:34:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/692012/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-ui","ref":"refs/changes/12/692012/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-ui refs/changes/12/692012/5"}}},"commit":{"parents":[{"commit":"835eac405b9777e2ff20802b99c187de716995e6","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/835eac405b9777e2ff20802b99c187de716995e6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-30 03:11:18.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-16 17:34:39.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nzaqar-ui is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\n\nChange-Id: Ia3f663e1271deb7a7aef6ee5fa1b17569fa5ce0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/f8a51b3e4ff4a0c851d23a8c247e38ac612e3718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/f8a51b3e4ff4a0c851d23a8c247e38ac612e3718"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
