)]}'
{"id":"openstack%2Fzaqar-ui~722980","triplet_id":"openstack%2Fzaqar-ui~master~I2fa66de2ad7655beeaeddcc3d95e3d3212bc1565","project":"openstack/zaqar-ui","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"I2fa66de2ad7655beeaeddcc3d95e3d3212bc1565","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-24 18:09:02.000000000","updated":"2020-04-27 00:07:16.000000000","submitted":"2020-04-27 00:04:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722980-1587945854992-593d93a0","meta_rev_id":"d03f4f082662f574b42d847119c08ce75de86d19","_number":722980,"virtual_id_number":722980,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 00:04:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":2,"date":"2020-04-26 23:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":1,"date":"2020-04-26 23:53:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 23:53:22.000000000","updated_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"reviewer":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"state":"REVIEWER"},{"updated":"2020-04-27 00:04:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8e6c4ff496ff90d6cd2e759be2dba2d69db3649","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-24 18:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d382e6734b0dd8b7d980f7b11c968761bb092447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 18:49:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d685a04002414cde927debd7dc74817e : SUCCESS in 8m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e42829111b54503848b6d62c48261e9 : SUCCESS in 6m 08s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/13824761a0e2417c9e1b9a941052cc04 : SUCCESS in 9m 51s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/32967c0b0ccb44df917597f69f53f640 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bccc93f37c2740569925d101b193922a : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a42d7e6f25b847c6a730c01e2cbdfdf1 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83e8a04b7dff46cbac6f4480a514a4f0 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02cce9dfc6b4678ba998c433b115f30 : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c82d0598b6966ff577d567bb31e471f14fd28518","author":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"date":"2020-04-26 23:53:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb4e6149e4ab8d021c17e669544397e4dc2ed5a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 23:53:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b90602af72522cbcd2f38f47b8a47a0d393caf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 00:04:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/147d9dfca455422e9d7f7c925f0eb627 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e98f296a26494db08b4accc7bc58ae62 : SUCCESS in 3m 56s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/3e21664d7bf54f428db1e3af0f889cc1 : SUCCESS in 5m 41s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/58096b0da9974a0da882702441be1234 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cdcb2e0f3374077bf91a57d3b9c6b05 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b3c2e9aeb7246d2ba0f0173a75ca180 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ea33eb61feb461b992a21cf85728fca : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b533f13a4c4dcf9682c6b75293268f : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"85e75266a9ae542a078f76fb560a7604ab865985","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 00:04:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4d3bb1d80cb868fc2fb606d94089618882249b96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 00:07:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe7509529c4416ea95559e4a4a6f3c6 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7ef2b451b62730971253105b39f7376d4550fdb","revisions":{"c7ef2b451b62730971253105b39f7376d4550fdb":{"kind":"REWORK","_number":1,"created":"2020-04-24 18:09:02.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/80/722980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-ui","ref":"refs/changes/80/722980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/80/722980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/80/722980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/80/722980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-ui refs/changes/80/722980/1"}}},"commit":{"parents":[{"commit":"6220cbd6d72bac368adc8cc523c1b46e4ddfe357","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/6220cbd6d72bac368adc8cc523c1b46e4ddfe357"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-24 18:08:59.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-24 18:08:59.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I2fa66de2ad7655beeaeddcc3d95e3d3212bc1565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/c7ef2b451b62730971253105b39f7376d4550fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/c7ef2b451b62730971253105b39f7376d4550fdb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
