)]}'
{"id":"openstack%2Fzaqar-ui~933321","triplet_id":"openstack%2Fzaqar-ui~master~I5df1ee190fb0738c46f206f301e8feb5b53dd51a","project":"openstack/zaqar-ui","branch":"master","hashtags":[],"change_id":"I5df1ee190fb0738c46f206f301e8feb5b53dd51a","subject":"Update supported python versions","status":"ABANDONED","created":"2024-10-24 14:17:01.000000000","updated":"2024-10-26 15:45:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"49f919dc30fcb798885be0cf344c705867675c86","_number":933321,"virtual_id_number":933321,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-24 14:34:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 14:34:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ab1269d163a1d6c74557629263a139df95075a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 14:17:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89a1a02f60f7dd4da0e9ffab66ab2c42d0d33f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 14:34:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8190d22d043f4e818fd153a0da114cfa\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c8a6bbfd05104f0d8b9629c27e165f83 : SUCCESS in 1m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77e3fd0b50464f87b0f8f5118faacc81 : SUCCESS in 3m 25s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e4c80600b8ec42fc930d1cfa8d655d30 : SUCCESS in 3m 48s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/ca2d98bd419849fc8be18b9df02bd320 : SUCCESS in 4m 01s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/8fa428f73286477587c566e29b8fd80e : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc8660e14bb4a439b8f5de8dcb0d8c4 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d539f2f053e4a3ca077224e0f2585b8 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c008e7e34f5943d9abc18e62dead2063 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61de5c8bf47842109466eed96a225815 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"49f919dc30fcb798885be0cf344c705867675c86","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 15:45:46.000000000","message":"Abandoned\n\nDuplicate of https://review.opendev.org/c/openstack/zaqar-ui/+/933508","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c44c381833d6211f72ff094c9132797afc77da63","revisions":{"c44c381833d6211f72ff094c9132797afc77da63":{"kind":"REWORK","_number":1,"created":"2024-10-24 14:17:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/21/933321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar-ui","ref":"refs/changes/21/933321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/21/933321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/21/933321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar-ui refs/changes/21/933321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar-ui refs/changes/21/933321/1"}}},"commit":{"parents":[{"commit":"8c2e94978b2f8fb3693410cc42597166170911cc","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/8c2e94978b2f8fb3693410cc42597166170911cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 14:16:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 14:16:58.000000000","tz":540},"subject":"Update supported python versions","message":"Update supported python versions\n\n... according to the version covered by current tested runtimes.\n\nChange-Id: I5df1ee190fb0738c46f206f301e8feb5b53dd51a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/c44c381833d6211f72ff094c9132797afc77da63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar-ui/commit/c44c381833d6211f72ff094c9132797afc77da63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
