)]}'
{"id":"openstack%2Fzaqar~1007158","triplet_id":"openstack%2Fzaqar~master~Ib7aadf8d6de396ac9f0a11d6fcfa44849b194bf4","project":"openstack/zaqar","branch":"master","topic":"bug/2164987","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-24 15:56:26.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":["CVE-2026-97404","ossa-2026-042"],"change_id":"Ib7aadf8d6de396ac9f0a11d6fcfa44849b194bf4","subject":"Fix authentication bypass by empty URL-Signature","status":"MERGED","created":"2026-09-24 14:12:24.000000000","updated":"2026-09-24 17:58:08.000000000","submitted":"2026-09-24 17:53:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007158-bug/2164987","meta_rev_id":"d66ca1aaefc7562b07bfe3af4fbf7cc0245090c4","_number":1007158,"virtual_id_number":1007158,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-24 17:53:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-24 15:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-24 15:15:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-24 14:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-24 15:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4197d9be65a93f73f7cd21545522935118d7375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 14:12:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38cf1339d4ece8dbcc4bff0a27f296f9c8792594","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 14:14:19.000000000","message":"Topic set to bug/2164987","accounts_in_message":[],"_revision_number":1},{"id":"e383bca665b74eaf1b7087f2e8e7a9872a2a281c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 14:35:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8803a932dda748428753e03c22b6cf3b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c16f0c3b2eb43389dad88aa2aa02d14 : RETRY_LIMIT in 3m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fb3aa94677de41e884c9fb6dc0b4b2f4 : SUCCESS in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87359de450449ebdd1551f6cfdb7f14d6fa668fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 15:04:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21e8ca98ad0044f698d3617ba378e9a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfe2d56c45384d6d8f75ea577ab95226 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ad6a98f7d3342a18e8fbc0f4573886c : SUCCESS in 13m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/99797e19fa60464ca0377750cad2f439 : SUCCESS in 12m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef4d48e68ee4836ac2d6b21bb2c1ffa : SUCCESS in 2m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad5cb11d9d4e4a0f95076b266b694a46 : SUCCESS in 2m 40s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/b58522485a8347428ca2dd28a4a0d567 : SUCCESS in 7m 37s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/634f2c5dd31e4d9da53f2f2d9053a423 : SUCCESS in 30m 25s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/6ef6aaf7b45a424ab415352c480b56f8 : SUCCESS in 19m 05s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/16e613221c5b437992d26c0a8d6eeeee : SUCCESS in 26m 09s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/9fc99edf491e4065a518302fac8258d4 : SUCCESS in 17m 17s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/0f8e42886b90427698f7ffce3174fa41 : FAILURE in 48m 52s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/51a9c334f5374589bdd53bc98a91ddd2 : SUCCESS in 26m 05s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/f8f81dd6b4a84954a5ef97ebd47aac38 : SUCCESS in 28m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"585f3e5151b3b7e4176be182474f2a8004050721","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 15:15:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"adbc0bdfca32abc4f13485208cd4b75a82094f94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 15:17:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b184bc8b22c3dd4ffa0d1c1e2abd556777d0c869","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-24 15:33:29.000000000","message":"Hashtag added: ossa-2026-042","accounts_in_message":[],"_revision_number":1},{"id":"215042dc61ea803ba2459fdc58693f35e9dd21ba","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-24 15:34:30.000000000","message":"Hashtag added: CVE-2026-97404","accounts_in_message":[],"_revision_number":1},{"id":"6f517c5fc08d615ef97f6d42626cbb0f31aa2543","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 15:55:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/111824eef658406ea5259e66231db14b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06ddecf1fb7b4846bfdaeb7c796bc83b : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cec5e613e57430d8c3e7f5a72590ddc : SUCCESS in 9m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5897fb6e81c441e2b55d0c0bc72f3ddc : SUCCESS in 17m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81fe584ba474ef68d021a2e6329001f : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc3ecbcf561141a9bc3e2b10911167ba : SUCCESS in 5m 10s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/15f00f2706bc4e279bcd1a3f5af9387e : SUCCESS in 8m 10s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/900aadd1273a409c874fa4b4d58f8582 : SUCCESS in 18m 38s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/a107d421f01b41c5b4a57dc40606f934 : FAILURE in 33m 19s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/1432f14f413f4b13a45451a3e238d2c7 : SUCCESS in 33m 04s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/26fcaad519104bc1b154a5764ca17ae4 : SUCCESS in 26m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3150dbd266a218cc06649f2ad2dcae42434c702b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 15:56:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56f4fe3692a912a31e92d3e67b3d493928d906a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 16:56:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/102ce41fa184435e85aedda80cab1976\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee2299cee6fc45ad9603381bb932bfb6 : RETRY_LIMIT in 2m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/209d1a5241ea4141941016b41bf564ea : SUCCESS in 17m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2796f17e6dc924bb42eea7407bdc00fcbc642ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:10:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bc22a438a194eff9731133089663c7b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3d64e6c62224a46b2f7d000c84ad9b1 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/959147b1d46b49e5bda1af3e1ad8c907 : SUCCESS in 10m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd4c3333942942a6808029c53766d0d0 : SUCCESS in 12m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b10f98fbf687424da6d791be5600904b : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c682db54d784343b947c59e37112653 : SUCCESS in 5m 27s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/383dc3caf98841b887e6fbd145635419 : SUCCESS in 7m 46s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/df84f34335604faab90036d8151bdc6b : SUCCESS in 40m 40s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/8377174f59f949e5912970f013600fc1 : SUCCESS in 29m 50s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/8f6b7f8673d8422f901d7b0ad2f4218b : SUCCESS in 16m 47s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/864836a65184499ea7af501b2be05d26 : SUCCESS in 32m 35s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/04542d6737ff4845859ff9b2b3fd8190 : FAILURE in 36m 57s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/2ba39881aa684778b39431a49c7bda16 : SUCCESS in 52m 53s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/e2163128bf2e4959a8b0f021b95942fb : SUCCESS in 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6565d7ce3924a84dce6a2f5274d0b42ea3063d66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:11:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6d45778710af3523562c8883144cbf5d415b960","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:53:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/896db2e637fe424fb708d9ea89511cdf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deb46ee99ae84465b69388746e25df75 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2713b2afeed47229306acba87079856 : SUCCESS in 8m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c070cdeb2253457cb1f6a08ccfc288f1 : SUCCESS in 14m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6850d7c37ef0471ab8533efdf66a246e : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7a1b710ecf64eadaaecf95dc2441923 : SUCCESS in 2m 50s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/d5d0ba75f9744dcbb4f13d0bf1616016 : SUCCESS in 6m 17s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/8af1ce3dcd33413396cd59195a6f2d95 : SUCCESS in 32m 40s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/7c25c6dc6f204698951517daf492894c : SUCCESS in 23m 09s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/297fa39d16684668a31cded80c3db1ee : SUCCESS in 18m 03s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/7572d41afdef4a19a4e2984bee69d0b3 : SUCCESS in 33m 15s","accounts_in_message":[],"_revision_number":1},{"id":"572e8220ebee760cbb68daee1aff8184228bd33e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:53:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d66ca1aaefc7562b07bfe3af4fbf7cc0245090c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:58:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bf2c7253d8449b5ad559996c884c3a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83edf96135144759e3a88a97eaf488d : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d470ad31fbb4987bf01b134d03a1277 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe1efeaf8abf40fe132207cce0cc14c470e98dee","revisions":{"fe1efeaf8abf40fe132207cce0cc14c470e98dee":{"kind":"REWORK","_number":1,"created":"2026-09-24 14:12:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/1007158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/58/1007158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/58/1007158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/58/1007158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/58/1007158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/58/1007158/1"}}},"commit":{"parents":[{"commit":"507446801264a12a70561caaac820a834c97efdd","subject":"Merge \"Get rid of marktime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/507446801264a12a70561caaac820a834c97efdd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 13:46:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 13:46:06.000000000","tz":540},"subject":"Fix authentication bypass by empty URL-Signature","message":"Fix authentication bypass by empty URL-Signature\n\nCloses-Bug: #2164987\nChange-Id: Ib7aadf8d6de396ac9f0a11d6fcfa44849b194bf4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/fe1efeaf8abf40fe132207cce0cc14c470e98dee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/fe1efeaf8abf40fe132207cce0cc14c470e98dee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
