)]}'
{"id":"openstack%2Fzaqar~1007162","triplet_id":"openstack%2Fzaqar~stable%2F2025.1~Ib7aadf8d6de396ac9f0a11d6fcfa44849b194bf4","project":"openstack/zaqar","branch":"stable/2025.1","topic":"bug/2164987","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-24 14:53:52.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":["CVE-2026-97404","ossa-2026-042"],"change_id":"Ib7aadf8d6de396ac9f0a11d6fcfa44849b194bf4","subject":"Fix authentication bypass by empty URL-Signature","status":"MERGED","created":"2026-09-24 14:14:00.000000000","updated":"2026-09-25 02:56:12.000000000","submitted":"2026-09-25 02:01:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007162-bug/2164987","meta_rev_id":"4bd248cca585cbfd8ff0aadbb213ffe1295295be","_number":1007162,"virtual_id_number":1007162,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 02:01:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-25 01:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-25 01:04:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-24 14:52:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9afeff88da2aec9e29cac38a7743edd1ffd337d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 14:14:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8b6e7946b8e96e55a66fb6f89286fe295a7b030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 14:52:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/474ed9d86e2d414cac0bea6da4fa91a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cff57a4b6c84ea7b20b6ca19d1e1c00 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33fbfa796ab84bc9b9dc232ff39fd56f : SUCCESS in 11m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5fd6738b64794bb79ad6b5b00ef3c957 : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25a81181a7e447d699b1e7386dd1be80 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e03007f81664489904d1323c3a91eb2 : SUCCESS in 4m 31s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/27c869ce2f1a4de08e5db75928ccd384 : SUCCESS in 10m 03s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/7f5ded989dc24d64a04425c1f5590b18 : SUCCESS in 30m 28s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/2ef669f82aaa4e9e840291f02fed6ccb : SUCCESS in 33m 29s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/c3b9e7176a6b4f3dae5e1114dbf6dddf : FAILURE in 22m 01s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/f7ed35d5d53040848ae5a4ec75c3a423 : SUCCESS in 26m 06s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/fdf29f60887c454c8403deb126d4cc3f : FAILURE in 17m 51s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/242daf12e48648d682d36c96b7e6a8df : FAILURE in 18m 39s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/36e2398475194eb28c6588b799ae0342 : FAILURE in 30m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe9697d3f75cf670da1f0444f084d8135277e169","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 14:53:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c34c6b7970248515b692c2808d3d7d1f30ccd57","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-24 15:33:29.000000000","message":"Hashtag added: ossa-2026-042","accounts_in_message":[],"_revision_number":1},{"id":"8e0c1be77bc5845a753580a29eac47c81e67ad1b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-24 15:34:30.000000000","message":"Hashtag added: CVE-2026-97404","accounts_in_message":[],"_revision_number":1},{"id":"481fceccda0b76b442108faad40863c1651418f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 15:36:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b44b72349b1e4253aea56b35e143d5e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/553c64bdc4384a8caadb2a71be9ceac4 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/351d9b7c13a94e849e85dfb19e1a406e : SUCCESS in 11m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b9b52fd774d4b80ba8dcda90ba89657 : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88dcabf80ec740b982c3e4afdfd35d68 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1aea3f94a67403a8b2aa5536b1bdd84 : SUCCESS in 5m 41s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/88319dbc93f7406daacc1b7e865cb10d : SUCCESS in 9m 52s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/7ca244d262394d929bd96a698366b7ab : SUCCESS in 25m 32s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/9bbf56e2a3774b25a6cc6e0499484916 : SUCCESS in 32m 24s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/ef33d748d4124ca698fac884a6a75308 : SUCCESS in 25m 30s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/db270afbacd747d5823bb4bf86dec395 : SUCCESS in 19m 42s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/0700225111d5473098afb86a144cf23b : FAILURE in 28m 32s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/6b140acd6ff845c8aa2f36ee29672368 : FAILURE in 20m 44s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/13641b57e136405d89bddad43252f96f : FAILURE in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d8e7ad691fb5ba921f91bfd23650a8e2033b29c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 01:04:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"087246c3591dbdcd1218026e0ed736b73f552966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 01:05:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0616c6507763f0e7fafc501a828669b1b60925e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 02:01:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72e96fe6aa464e0bb8237ec5998ba75d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4e5aa94b1f945a584b9b262027756f3 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfaf17cfb2444cd19b06c68c932c79a3 : SUCCESS in 12m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b69aa49c11547b0bd9bb9d2442d76c9 : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e51ca9fbf1d4b6e856c2c4efd1ad51a : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8715a86a3b840178b239deea4ae42ef : SUCCESS in 4m 45s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/d7c6ed2283a849ce8655921247f37909 : SUCCESS in 12m 30s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/fc0eb25638f2441ca43c43da183e3aed : SUCCESS in 31m 56s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/920c22b1d1674b339ac1efbd67bd01c1 : SUCCESS in 33m 07s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/6b0700f047dc4b60b075ef734ed59b38 : SUCCESS in 33m 07s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/e92551f84c56421c92fc2cfe0f15c6fd : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2a0923f2eb0ee4808d56c9205aec83e06188f682","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 02:01:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4bd248cca585cbfd8ff0aadbb213ffe1295295be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 02:56:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5167a388e5d44b6282c8dd51fa9cbe0c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0bf3f7eec7f4e5f9a75d5124079dadf : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80a74e8822b64128a5b1d26bc28c5461 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"218c388a1efaa7d64470dcd7193d057613c3e6bc","revisions":{"218c388a1efaa7d64470dcd7193d057613c3e6bc":{"kind":"REWORK","_number":1,"created":"2026-09-24 14:14:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/1007162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/62/1007162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/62/1007162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/62/1007162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/62/1007162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/62/1007162/1"}}},"commit":{"parents":[{"commit":"07b283de721b30313b5c5625d63eb875d7a0ec26","subject":"Do not bypass authentication for requests with EXTRA-SPEC header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/07b283de721b30313b5c5625d63eb875d7a0ec26"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 13:46:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 14:13:52.000000000","tz":540},"subject":"Fix authentication bypass by empty URL-Signature","message":"Fix authentication bypass by empty URL-Signature\n\nCloses-Bug: #2164987\nChange-Id: Ib7aadf8d6de396ac9f0a11d6fcfa44849b194bf4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit fe1efeaf8abf40fe132207cce0cc14c470e98dee)\n(cherry picked from commit e7d8236443b7295c996a37f91bc87270a0f1a994)\n(cherry picked from commit 9b3313a78cc81f90c065c44cf7d1b9fe060d48d9)\n(cherry picked from commit 7a18ffc9d468a5fd8a58fc69468a0003d47ef4cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/218c388a1efaa7d64470dcd7193d057613c3e6bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/218c388a1efaa7d64470dcd7193d057613c3e6bc"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
