)]}'
{"id":"openstack%2Fzaqar~1007346","triplet_id":"openstack%2Fzaqar~stable%2F2025.1~Ie0d982b78612a149a602b29b9573cc50a4ed3668","project":"openstack/zaqar","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-25 04:33:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ie0d982b78612a149a602b29b9573cc50a4ed3668","subject":"Add test case to check empty signature header","status":"MERGED","created":"2026-09-25 02:12:10.000000000","updated":"2026-09-25 05:10:16.000000000","submitted":"2026-09-25 05:08:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007346","cherry_pick_of_change":1007345,"cherry_pick_of_patch_set":1,"meta_rev_id":"b41d0ea00fce493ee5c01b0cc6f1685cefbbbd27","_number":1007346,"virtual_id_number":1007346,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 05:08:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-25 04:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-25 04:33:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 03:30:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecc71ca119a8fe3230b20731fa508ad4f5fd03ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 02:12:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"0ce20c49d584eb05c146f68d73a749fb6fbbc8ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 03:30:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92fca954ff484f458df414ae69ae0363\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88bd556aabad42149edc476bbaef072d : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fcea3bf07d24c4cabead2162e941222 : SUCCESS in 11m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/101113cdb3b246d58c24c15215dcc49b : FAILURE in 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52394727cf9c4dff9ee9de218454d259 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59b35fe716b745908aa29abaf49ebe8f : SUCCESS in 5m 31s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/fa3738b2be5749b795852953374ad42a : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f47ad2690ef3bf453e6e504b5e56ea48983ee8e4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 04:33:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6e150194661b24fc062d6bddf2e4a39dbb4488a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 04:52:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7a6048984954fc59c14130839f907a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f56a1759635f4839b6a4b29f50966dfc : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea42b2632d5441339ee9e4aae353f486 : SUCCESS in 12m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c45491edb06a4e33a31b5f43a1af42c0 : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ce51933aae4a2194ace9f232e0cbff : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aefe086abd794da396273ad26d9a8591 : SUCCESS in 5m 17s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/7ef1003f7106434ca5412ec4b2918b9a : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":1},{"id":"73f28cac3a290100f5a8b46e02e2f4d37cc8d1e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 04:53:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e4c24af74dedf45b8165905fc7e371bd378637c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 05:08:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9db0bc892ab041a9a82810599b9c43e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca7045d7f82489aace018b147580b6a : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66346f9a44a3418d828ae2167a9bbf29 : SUCCESS in 11m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/376f210db121490191dae9ec14d30a1e : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b873b9db124f929e61ce25b5aa507b : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07763ddbaa46467697acd0979349e2c7 : SUCCESS in 4m 36s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/d439c4b40bc94400a10c156a6978b360 : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a520290636924b55a281badc1410a88ab154b23b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 05:08:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b41d0ea00fce493ee5c01b0cc6f1685cefbbbd27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 05:10:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e310ed73c14e66880cd27ad350223d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d39baba1ef1c4e59ae08790b10dd1ab6 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be9c2e48c5799a35e3e2eacf1892e80e3ac6e37b","revisions":{"be9c2e48c5799a35e3e2eacf1892e80e3ac6e37b":{"kind":"REWORK","_number":1,"created":"2026-09-25 02:12:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/1007346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/46/1007346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/46/1007346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/46/1007346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/46/1007346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/46/1007346/1"}}},"commit":{"parents":[{"commit":"218c388a1efaa7d64470dcd7193d057613c3e6bc","subject":"Fix authentication bypass by empty URL-Signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/218c388a1efaa7d64470dcd7193d057613c3e6bc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 14:20:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 02:12:10.000000000","tz":0},"subject":"Add test case to check empty signature header","message":"Add test case to check empty signature header\n\nRelated-Bug: #2164987\nChange-Id: Ie0d982b78612a149a602b29b9573cc50a4ed3668\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit f9546412afd5c5e5734928638fc0ffde5b457182)\n(cherry picked from commit eff94b00ee292f1790c9bc5105e7f7c73292edb7)\n(cherry picked from commit b30f5cb937a9fbdeb469fd6bde23741de29565c5)\n(cherry picked from commit d7d24bdebef2db0de987866b4a66ea12cbd28a20)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/be9c2e48c5799a35e3e2eacf1892e80e3ac6e37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/be9c2e48c5799a35e3e2eacf1892e80e3ac6e37b"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"7a18ffc9d468a5fd8a58fc69468a0003d47ef4cf","ours":"218c388a1efaa7d64470dcd7193d057613c3e6bc","theirs":"d7d24bdebef2db0de987866b4a66ea12cbd28a20","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
