)]}'
{"id":"openstack%2Fzaqar~800151","triplet_id":"openstack%2Fzaqar~master~Ifb8d4f589c2fb8fa2adc0372d1d54077afe5b32d","project":"openstack/zaqar","branch":"master","hashtags":[],"change_id":"Ifb8d4f589c2fb8fa2adc0372d1d54077afe5b32d","subject":"Changed minversion in tox to 3.18.0","status":"MERGED","created":"2021-07-09 02:40:49.000000000","updated":"2021-08-19 01:46:28.000000000","submitted":"2021-08-19 01:44:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800151","meta_rev_id":"e9cf718649a574886dfce184b0a6a1fb67077631","_number":800151,"virtual_id_number":800151,"owner":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 01:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-19 00:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-19 00:59:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 04:02:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 00:59:58.000000000","updated_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"REVIEWER"}],"messages":[{"id":"22fb7e17f1919ccf788e67d8fa8e3f64f79c7041","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-07-09 02:40:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adb0e30e44f3f7c25361d1d6418cd8a7e4ffc1ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 04:02:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1c7e196cb47142cbada83220d05484c9 : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6b535f045bd4052aaf7a43701545bea : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277ced9f2911415cba0c686d2937703b : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9183144da457454f8df6e00f943c6a7a : SUCCESS in 11m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79b0639eb4934053a75ef06be664e53e : SUCCESS in 11m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a600a2d849e4ef69b740f83e7e6f5a5 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8afe80f7a2894401aec51fc13ea636d1 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63de0c8304db4121a6a5e75895e37c5a : SUCCESS in 5m 31s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/a34d1ef244574bd1b3b1c18168644d44 : SUCCESS in 12m 21s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/88c1402bd4e8461784e08282677a796d : SUCCESS in 34m 32s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/62653c7981494f1389cac939bcc9bdb1 : SUCCESS in 35m 21s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/74a7d2e5411f4f0d8126bc83636f6a27 : SUCCESS in 35m 26s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/8a7828d159734de1a24774e37171d602 : SUCCESS in 35m 57s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/d0e60209aac44cde89440b6a7d4d4408 : SUCCESS in 58m 47s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/7765ff819b0b4bd49255d64e51bb0d70 : SUCCESS in 1h 15m 04s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/a879f241afac4995b8e4eacae7924496 : SUCCESS in 1h 07m 19s (non-voting)\n- zaqar-rally https://zuul.opendev.org/t/openstack/build/444646dcc3674c2b8da2a646232769e5 : SUCCESS in 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0c56e0dcff585e210f5b3543412abcd1dc8e304","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2021-08-19 00:59:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"501f78a1b0efa6777d6de55b28390db4b061131a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 01:00:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4cb9256944fbeb11c74562a7aa44d8eb19ba5cbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 01:44:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6ffc7fbd2bf34101b2d1b95c795e6548 : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bd98fae98124e25a142aabb85906d10 : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e50ed66948fb401fa593436564e375c9 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a68a97ca05bc4c9691d167e3cf67dd13 : SUCCESS in 11m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afe9b929b7c14851ae384028f9896e4e : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab910d814b5f402fb6a73a31f170e24f : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fb4e154e97342e4be2e3f066fe008be : SUCCESS in 5m 18s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/4e5e8fc798a1410c97beafe3902e192b : SUCCESS in 12m 36s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/7e3900629dca435787ca0a38b3b0a91b : SUCCESS in 34m 22s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/23f5a248c89b459da6d60f27f3d8d262 : SUCCESS in 35m 31s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/94b7ae1ad3bb483e8ec99d4f8883f1bc : SUCCESS in 33m 56s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/1c8877dbf8974fe29a2edae61847cd4e : SUCCESS in 27m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d0829e5f98c7af332619279f59141b5134508eeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 01:44:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e9cf718649a574886dfce184b0a6a1fb67077631","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 01:46:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ef0cb01556ca4d2c99ab062a2bd3a0a3 : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/232d11ec26c54f7aa951cd442099c1bb : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6e83d5bb8764b16ac15e40bb3076892 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8dae90a98cc37453cf1da900673eb67a4b28a047","revisions":{"8dae90a98cc37453cf1da900673eb67a4b28a047":{"kind":"REWORK","_number":1,"created":"2021-07-09 02:40:49.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/51/800151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/51/800151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/51/800151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/51/800151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/51/800151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/51/800151/1"}}},"commit":{"parents":[{"commit":"fad3fa1416f2611bd883f3c46ddb2225f68e57e6","subject":"[community goal] Update contributor documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/fad3fa1416f2611bd883f3c46ddb2225f68e57e6"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-07-09 02:40:15.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-07-09 02:40:44.000000000","tz":480},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: Ifb8d4f589c2fb8fa2adc0372d1d54077afe5b32d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/8dae90a98cc37453cf1da900673eb67a4b28a047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/8dae90a98cc37453cf1da900673eb67a4b28a047"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
