)]}'
{"id":"openstack%2Fzaqar~809652","triplet_id":"openstack%2Fzaqar~stable%2Fqueens~Id434fc1f2b7746d9d25991d38a7dc0a8d4f64986","project":"openstack/zaqar","branch":"stable/queens","topic":"remove-c7-jobs-stable/queens","hashtags":[],"change_id":"Id434fc1f2b7746d9d25991d38a7dc0a8d4f64986","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-17 12:37:17.000000000","updated":"2022-12-05 18:43:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809641,"cherry_pick_of_patch_set":1,"meta_rev_id":"14bc07a5c4d72115553493720a18191615d61e6a","_number":809652,"virtual_id_number":809652,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-17 13:21:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 13:21:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3cb6536822bbf61014b6abe3bfba8cba3ffe9b18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 12:37:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"666f45121f1efe4c23cd330261a9b13b5bfd6249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 13:21:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c145b7b75bc546e9bb25e6d52b512a34 : SUCCESS in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91f0645bba764b768f2c40abd21e89a0 : FAILURE in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3670a1730b9c42b9819f4f30980ff4bb : SUCCESS in 14m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/66d7749c0ab44d78b6c232a73edb8558 : SUCCESS in 6m 07s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/358a752b253d4c758bf770eaac7ccae6 : TIMED_OUT in 31m 53s\n- zaqar-tempest-redis https://zuul.opendev.org/t/openstack/build/848e0f83d925443b8a2e6c9641a9afa1 : FAILURE in 24m 58s\n- zaqar-tempest-swift https://zuul.opendev.org/t/openstack/build/86312f71040845b0b410d53ce4ebad48 : FAILURE in 29m 17s\n- zaqar-tempest-mongodb https://zuul.opendev.org/t/openstack/build/8196499b881e432f9893eb2d7ffbfad1 : FAILURE in 24m 12s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/21f58eeb729e419ab417caf34ad1ec6c : FAILURE in 24m 21s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/c0102cf8db5d4458874ab17dc0013e5c : FAILURE in 27m 26s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/d1eaba254e0d489b9b2294bf50488118 : FAILURE in 26m 37s (non-voting)\n\nWarning:\n  Comments left for invalid file samples/python-zaqarclient/receive_message/zaqar_sample.py\n  Comments left for invalid file zaqar/tests/unit/storage/test_impl_sqlalchemy.py\n  Comments left for invalid file zaqar/storage/redis/flavors.py\n  Comments left for invalid file samples/python-zaqarclient/send_message/zaqar_sample.py","accounts_in_message":[],"_revision_number":1},{"id":"14bc07a5c4d72115553493720a18191615d61e6a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-05 18:43:10.000000000","message":"Abandoned\n\nThis branch of the project has been moved to End of Life. In order to be able to delete the branch open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"054de6b31275c6d7dabddf268ad5e9fb810ab156","revisions":{"054de6b31275c6d7dabddf268ad5e9fb810ab156":{"kind":"REWORK","_number":1,"created":"2021-09-17 12:37:17.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/52/809652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/52/809652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/52/809652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/52/809652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/52/809652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/52/809652/1"}}},"commit":{"parents":[{"commit":"09fad0b85d9c93e1f70587d0b57a7243973769a9","subject":"Import legacy-rally-dsvm-zaqar-zaqar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/09fad0b85d9c93e1f70587d0b57a7243973769a9"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 12:36:09.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 12:37:17.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id434fc1f2b7746d9d25991d38a7dc0a8d4f64986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/054de6b31275c6d7dabddf268ad5e9fb810ab156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/054de6b31275c6d7dabddf268ad5e9fb810ab156"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
