)]}'
{"id":"openstack%2Fzaqar~809653","triplet_id":"openstack%2Fzaqar~stable%2Fpike~Id434fc1f2b7746d9d25991d38a7dc0a8d4f64986","project":"openstack/zaqar","branch":"stable/pike","topic":"remove-c7-jobs-stable/queens-stable/pike","hashtags":[],"change_id":"Id434fc1f2b7746d9d25991d38a7dc0a8d4f64986","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-17 12:37:29.000000000","updated":"2022-08-01 16:12:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809652,"cherry_pick_of_patch_set":1,"meta_rev_id":"c36631303b79df718da5701c053437d334de90e8","_number":809653,"virtual_id_number":809653,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-17 13:04:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 13:04:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31f86cc6a59d6a8d42b75333a620090c43b9812a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 12:37:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/queens.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"219fea9d2da65831f00b381ac493476bd45ad5dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 12:38:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"df43fc48a8bd03d8518fda5672d4c16907c7be7c","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 12:38:36.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"b106005f7dce74f38051dc15cbdc1a8acc50efb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 13:04:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/04b4049998904d7a8a5727f58ee8d8de : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6697b6d6631c4824ae185439e87fb1f6 : FAILURE in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b86834f867ad4a98bbd6cb9ccba346f8 : SUCCESS in 13m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f21077544aad42edaefd3cf6f2976f71 : SUCCESS in 7m 42s\n\nWarning:\n  Comments left for invalid file zaqar/tests/unit/storage/test_impl_sqlalchemy.py\n  Comments left for invalid file samples/python-zaqarclient/receive_message/zaqar_sample.py\n  Comments left for invalid file samples/python-zaqarclient/send_message/zaqar_sample.py","accounts_in_message":[],"_revision_number":2},{"id":"c36631303b79df718da5701c053437d334de90e8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-01 16:12:00.000000000","message":"Abandoned\n\nstable/pike has transitioned to End of Life for zaqar, open patches need to be abandoned in order to be able to delete the branch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a4f79d191b089087bf12c7fc364e55883b40c1b0","revisions":{"7a965151193029427b2bb84be5ac0f3d7aac19c0":{"kind":"REWORK","_number":1,"created":"2021-09-17 12:37:29.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/53/809653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/53/809653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/53/809653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/53/809653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/53/809653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/53/809653/1"}}},"commit":{"parents":[{"commit":"251e17990cd06ef356eadd416f1dd6a58493a14c","subject":"Import legacy-rally-dsvm-zaqar-zaqar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/251e17990cd06ef356eadd416f1dd6a58493a14c"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 12:36:09.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 12:37:29.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id434fc1f2b7746d9d25991d38a7dc0a8d4f64986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/7a965151193029427b2bb84be5ac0f3d7aac19c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/7a965151193029427b2bb84be5ac0f3d7aac19c0"}]},"branch":"refs/heads/stable/pike"},"a4f79d191b089087bf12c7fc364e55883b40c1b0":{"kind":"REWORK","_number":2,"created":"2021-09-17 12:38:25.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/53/809653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/53/809653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/53/809653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/53/809653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/53/809653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/53/809653/2"}}},"commit":{"parents":[{"commit":"251e17990cd06ef356eadd416f1dd6a58493a14c","subject":"Import legacy-rally-dsvm-zaqar-zaqar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/251e17990cd06ef356eadd416f1dd6a58493a14c"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 12:36:09.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 12:38:21.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id434fc1f2b7746d9d25991d38a7dc0a8d4f64986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/a4f79d191b089087bf12c7fc364e55883b40c1b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/a4f79d191b089087bf12c7fc364e55883b40c1b0"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
