)]}'
{"id":"openstack%2Fzaqar~833321","triplet_id":"openstack%2Fzaqar~stable%2Fyoga~I73a654d8bb6e056ad6ab0277a36eeaacdcd333b7","project":"openstack/zaqar","branch":"stable/yoga","topic":"create-yoga","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-19 17:06:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I73a654d8bb6e056ad6ab0277a36eeaacdcd333b7","subject":"Update .gitreview for stable/yoga","status":"MERGED","created":"2022-03-11 11:54:53.000000000","updated":"2023-06-19 17:07:19.000000000","submitted":"2023-06-19 17:06:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833321-create-yoga","meta_rev_id":"10427cee12cbec3b558c5713d661920add7afdd6","_number":833321,"virtual_id_number":833321,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-19 17:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-06-07 20:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-06-07 20:49:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-11 13:26:00.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-03-11 14:17:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 01:22:08.000000000","updated_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"CC"},{"updated":"2023-06-07 20:49:02.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"a662263a72be87cb81b6bc7dc8dcb4764797c53c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-03-11 11:54:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"891064d98b65d36d6b7d5f17109bdc98e2ec6e6e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-11 13:26:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05da1b7efeee6e463c80913503c393fd1edc07ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 14:17:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0855108a2878408498fe05b0323640b6 : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe6de78ee2842d6a79aff09a75f1ece : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/869363a3832a4d459acebfb31477f8fb : SUCCESS in 13m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e0a3603e15d4fc8a7934468b45e7f9a : SUCCESS in 12m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cfcee95e00b481da3f296aaeef47042 : SUCCESS in 6m 17s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/2191b986aad2475f81320afe89821509 : SUCCESS in 14m 55s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/abc132f9051d41208fd4f1663bdf7c43 : FAILURE in 14m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7a20277495585fb85df04d70bb0ae2a8d8b8fe46","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2022-03-15 01:22:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0dc94fca1d5385b03cbdbf6764c6ee756b815b89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 01:57:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaf827ca766a4e809b33c87b0473b46f : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f303a07bc9974a10b38869dd5f298fa7 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/434e61a653ee4e1fba6c82fa28bf3228 : SUCCESS in 11m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3639719ba8649bc858a238b059b2f9c : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6da0837fb94a31839429fdd3418abc : SUCCESS in 6m 08s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/16a47a8cb1a146208368e08840e64dfb : SUCCESS in 13m 38s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/0261afb050bd47f3a4fba331e9b750b8 : FAILURE in 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"703aa29461d45e2f48114d4e374a4600ef98a4d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-06 16:51:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9384544ae4277f1d8da63abcaae59e623f638848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 18:02:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3a12ab3b96042e59d6c8833f5071979\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0637502044b543b69ce83e84963167b8 : FAILURE in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc6b10e08faf447d9b499429529eed24 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5a690546d3c44e1bc1ce98628ae749f : SUCCESS in 11m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e618bc763757409989c984cf08935c05 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de1fd0d1a4554d48ac499cbeaab43a3b : SUCCESS in 4m 59s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/e5079b83bacf44609c13e4e9bbca1e14 : SUCCESS in 11m 32s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/3084883ce8f3452aad3302538fdd724d : SUCCESS in 43m 33s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/36038dff37724d16b66604a3fda91986 : SUCCESS in 29m 41s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/4d32e9df54ce47f0b28609dd8ab0d103 : SUCCESS in 46m 50s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/c05f6f513ea44fd5ba9b11fedab6c2c1 : SUCCESS in 28m 50s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/1ea157dc100740298e1f4f4e2efe5f0b : SUCCESS in 42m 27s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/a3223ca430a149989403a421284ffedf : SUCCESS in 1h 03m 53s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/2dc3fa2dce3d4fd0bad07e021482b6fa : SUCCESS in 44m 33s (non-voting)\n- zaqar-rally https://zuul.opendev.org/t/openstack/build/567c257caab341f5b2bdaec77a42963e : SUCCESS in 26m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"96883e24373259626c9e6d3fcbf6f64d0a030347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-07 18:54:42.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b87d8518911a82d914f9ae5ac005e41fd7811ec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 20:12:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eec782d6ea264738abe3471e606a4ca4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c57481be2b40b8a1d666f4a208e9df : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c59c887a7064abb9f2de22794d3a297 : SUCCESS in 11m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5bfe4bd58e44e0fadda2c15bdffb51d : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c1cb20d9274cdd8f87a848e1b0a41b : SUCCESS in 4m 59s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/c484ff3f4f914b8c828b037728b6a627 : SUCCESS in 11m 40s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/ebc339fc54b949e1a544e75012988b88 : SUCCESS in 25m 17s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/095c14e5a8d6409ca4a121b6cffc696e : SUCCESS in 30m 07s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/37b9191aa299415fa4918ff62fafb69b : SUCCESS in 40m 24s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/f676df1732054a8ab2efcd88826680e2 : SUCCESS in 40m 07s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/2781a3c4b8084306819b469644ddd814 : SUCCESS in 45m 51s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/daa05aa931d54e0f90c46b0bcdcb9e9e : SUCCESS in 47m 13s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/8b572ea9d09a46c68b4b00f23249b8b7 : SUCCESS in 1h 10m 50s (non-voting)\n- zaqar-rally https://zuul.opendev.org/t/openstack/build/6f5edbedc8eb4012923c482e75c3979d : SUCCESS in 40m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"17d930d3bbccc3b6dcb8a80527b600f782d1c034","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-06-07 20:49:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ffdedd3855053da86aab23d7cb267b1ac947d146","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 16:14:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0bf738f4ffd3ed4518f34d1c9f2e27824d432ba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 17:06:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f0e5e84ebc433182add71c497c470a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ca2e1adc3eb401f95a366a77f6d9014 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3924590b20584b8fa09f777aadb0bdd3 : SUCCESS in 11m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54db1495d3b64e6f952347a11558a986 : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e66eb452e2ae445788c14549c78ada10 : SUCCESS in 5m 24s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/a17629e13cb84890998ad4080d9518ed : SUCCESS in 11m 39s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/1f0362b150194006b011e88685516fa9 : SUCCESS in 42m 41s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/6fc053de78f94a2d8fab2ce337514369 : SUCCESS in 23m 14s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/b7e7135cd0ae494694b8afcfec152d58 : SUCCESS in 44m 51s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/1d19e5e082ab4ad299ac9658d4f1bdf2 : SUCCESS in 44m 40s","accounts_in_message":[],"_revision_number":3},{"id":"781b499d427d18197ee6c9519c9c55ed3c1eb02b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 17:06:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"10427cee12cbec3b558c5713d661920add7afdd6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 17:07:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8975a4edab9f4b83984fa70ce4e85c78\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16bc160fdc74490a94b2c73e14615d09 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e3e814c695b294e9f6897fe6aeeeb4436709cc3e","revisions":{"5b697fd8bceba0ee9de9449f3b37f30733390f5a":{"kind":"REWORK","_number":1,"created":"2022-03-11 11:54:53.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/21/833321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/21/833321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/21/833321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/21/833321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/21/833321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/21/833321/1"}}},"commit":{"parents":[{"commit":"14c3ffbe8e60835fdffe1b32638f01307b1a4875","subject":"Merge \"[Part 10] Remove all usage of six library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/14c3ffbe8e60835fdffe1b32638f01307b1a4875"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-11 11:54:53.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-11 11:54:53.000000000","tz":0},"subject":"Update .gitreview for stable/yoga","message":"Update .gitreview for stable/yoga\n\nChange-Id: I73a654d8bb6e056ad6ab0277a36eeaacdcd333b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/5b697fd8bceba0ee9de9449f3b37f30733390f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/5b697fd8bceba0ee9de9449f3b37f30733390f5a"}]},"branch":"refs/heads/stable/yoga"},"77cf35749f889c5d9b1684a7071270e9be08f7e1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-06 16:51:39.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/21/833321/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/21/833321/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/21/833321/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/21/833321/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/21/833321/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/21/833321/2"}}},"commit":{"parents":[{"commit":"84395bde183eeab7686afbc05e0ed68abe7893f8","subject":"Update deprecated zuul syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/84395bde183eeab7686afbc05e0ed68abe7893f8"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-11 11:54:53.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-06-06 16:51:39.000000000","tz":0},"subject":"Update .gitreview for stable/yoga","message":"Update .gitreview for stable/yoga\n\nChange-Id: I73a654d8bb6e056ad6ab0277a36eeaacdcd333b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/77cf35749f889c5d9b1684a7071270e9be08f7e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/77cf35749f889c5d9b1684a7071270e9be08f7e1"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"e3e814c695b294e9f6897fe6aeeeb4436709cc3e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-06-07 18:54:42.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/21/833321/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/21/833321/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/21/833321/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/21/833321/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/21/833321/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/21/833321/3"}}},"commit":{"parents":[{"commit":"6ea92894735adfa63bb8cccc91e4268415a723d8","subject":"Update deprecated zuul syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/6ea92894735adfa63bb8cccc91e4268415a723d8"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-11 11:54:53.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-06-07 18:54:42.000000000","tz":0},"subject":"Update .gitreview for stable/yoga","message":"Update .gitreview for stable/yoga\n\nChange-Id: I73a654d8bb6e056ad6ab0277a36eeaacdcd333b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/e3e814c695b294e9f6897fe6aeeeb4436709cc3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/e3e814c695b294e9f6897fe6aeeeb4436709cc3e"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
