)]}'
{"id":"openstack%2Fzaqar~853044","triplet_id":"openstack%2Fzaqar~master~Ib26f5d6b8415fdf454783b7120434ffd43ed5559","project":"openstack/zaqar","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2022-09-16 07:51:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib26f5d6b8415fdf454783b7120434ffd43ed5559","subject":"remove unicode prefix from code","status":"MERGED","created":"2022-08-15 02:05:03.000000000","updated":"2022-09-16 07:53:11.000000000","submitted":"2022-09-16 07:51:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853044","meta_rev_id":"27f5e9da965fde138d6de06ba176dfae5e270113","_number":853044,"virtual_id_number":853044,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-16 07:51:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-16 07:02:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-16 07:02:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-15 03:30:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-16 07:02:24.000000000","updated_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"REVIEWER"}],"messages":[{"id":"d83ffd078e772bc08959ba189fcd5ab6681b5cc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-15 02:05:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a748ed5b1ddba2abc60f120a74c9585017f5880e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-15 02:06:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd5ba2cfec0a7fb4fb746dad93621e255f0537ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 03:30:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5267995e4da84b0e9d77da69e068890d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4733226e202c44f69ae6e1225ea04686 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2acc0ea1064473db779009a66619b06 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c25fb1200f546649dcb82c814d1861c : SUCCESS in 5m 52s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/d16148f632ce4014b879efd7ed7b666f : SUCCESS in 11m 27s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/6151478f679b411eb2e088f8c0530078 : SUCCESS in 45m 13s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/59b86e3bfda04396a3eb13b02b5b675c : SUCCESS in 43m 29s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/f123702aada2411fb2d0de6d4a29b0e0 : SUCCESS in 43m 27s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/d8fa6b51f85e400aad660e96ca18ae53 : SUCCESS in 45m 05s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/f5ab6b9ca898458ba582b3875a4a8ae5 : SUCCESS in 1h 00m 26s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/ba367c04642948ba93f60ce79d4f2ffe : SUCCESS in 58m 30s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/43d807dab31b4f669a24737e160b9261 : SUCCESS in 1h 11m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5971f8a32a792bcd669f594a3c52fb78dac24734","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2022-09-16 07:02:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0663ca7ecd2b7d3439d52c8dc6ac6232fabe9585","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 07:03:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b15bf942722b9c924fb9c727779c000203935b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 07:51:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2bf035a36bd4e3e97c1ce042cdacbd2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30eef21485014e0c944008261140aa70 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c8bf60789484457bf2c4d16237588cd : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8df0facdd2074d1e90733de07e6d1f48 : SUCCESS in 5m 14s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/add425b76d094650afde9c70d5eb0336 : SUCCESS in 12m 11s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/64f8fcb5795343f79306f411bb0e1d0f : SUCCESS in 39m 58s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/57eefcdc6fde4a63ba1813c5fe9f4cf3 : SUCCESS in 40m 30s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/e19ac7440d554186a05ea34a05b1b727 : SUCCESS in 41m 44s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/2616d1c242ab414aa8ae80fa37976fb6 : SUCCESS in 41m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b4a092b8869e3e49a0d1a03c42ae3d7b61013aff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 07:51:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"27f5e9da965fde138d6de06ba176dfae5e270113","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 07:53:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9395b97540424dc1ac7d07868ac0eea9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c05a7c96cc547ae92fff96bb720b9a4 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9151fd19c3de4e2188b4c3c897d7e06d : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4a035775a099e5603a062712529f0e5745f45420","revisions":{"d4cb8b5e12628290d2772937979cb75c99051fbe":{"kind":"REWORK","_number":1,"created":"2022-08-15 02:05:03.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/44/853044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/44/853044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/44/853044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/44/853044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/44/853044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/44/853044/1"}}},"commit":{"parents":[{"commit":"cdf45da5553f4daed20dc6bd8800360130c75f55","subject":"Fix API DOC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/cdf45da5553f4daed20dc6bd8800360130c75f55"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 02:04:35.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 02:04:40.000000000","tz":480},"subject":"remove unicode prefix from code","message":"remove unicode prefix from code\n\nChange-Id: Ib26f5d6b8415fdf454783b7120434ffd43ed5559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/d4cb8b5e12628290d2772937979cb75c99051fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/d4cb8b5e12628290d2772937979cb75c99051fbe"}]},"branch":"refs/heads/master"},"4a035775a099e5603a062712529f0e5745f45420":{"kind":"REWORK","_number":2,"created":"2022-08-15 02:06:10.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/44/853044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/44/853044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/44/853044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/44/853044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/44/853044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/44/853044/2"}}},"commit":{"parents":[{"commit":"cdf45da5553f4daed20dc6bd8800360130c75f55","subject":"Fix API DOC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/cdf45da5553f4daed20dc6bd8800360130c75f55"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 02:04:35.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 02:05:54.000000000","tz":480},"subject":"remove unicode prefix from code","message":"remove unicode prefix from code\n\nChange-Id: Ib26f5d6b8415fdf454783b7120434ffd43ed5559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/4a035775a099e5603a062712529f0e5745f45420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/4a035775a099e5603a062712529f0e5745f45420"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
