)]}'
{"id":"openstack%2Fzaqar~853045","triplet_id":"openstack%2Fzaqar~master~I132e2e4f2c71217bc42b28b53a52a1be6d3b19b4","project":"openstack/zaqar","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2022-09-16 08:44:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I132e2e4f2c71217bc42b28b53a52a1be6d3b19b4","subject":"Cleanup py27 support","status":"MERGED","created":"2022-08-15 02:07:54.000000000","updated":"2022-09-16 08:46:23.000000000","submitted":"2022-09-16 08:44:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853045","meta_rev_id":"538b4ae18ab248c1ae2fd751f903c7b66db4a24d","_number":853045,"virtual_id_number":853045,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-16 08:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-16 07:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-16 07:03:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-15 02:50:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-15 03:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-16 07:03:08.000000000","updated_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"REVIEWER"}],"messages":[{"id":"e76f9c49a563d96d892b3d3ff2bc47888cb55006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-15 02:07:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87ac4b79ba7f5410acd3b1fd5cde92c05836d3f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 02:50:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d42a874cb6a4c3eb016ef6fa1ad8a4d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae4c66407c3e46d0aadb144401536add : SUCCESS in 19m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1f7dd2f62ba475eb3850374fa209cee : SUCCESS in 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa72f4a2050dd99f3af6de111471c47531904dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 03:51:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/978dca44e2b24925932c6a05bc8b9821\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c2821022f2144d8b80a1a32da3fbd2e5 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb4fdd004e6e415d9de44abfb55a5e2f : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/068dfada27fb4f5f90516d3cbfc48600 : SUCCESS in 11m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cc6d11694ed4d8d8e5c9b8dc4b285e1 : SUCCESS in 11m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bf6b754ad5a4b37ba1b0963bf97dff9 : RETRY_LIMIT in 3m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff634f16afa84e1e8d63682478fa2487 : SUCCESS in 5m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/a9d21b2f34784d7bb405a7f37ddf0993 : SUCCESS in 1h 11m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4781605270e4473969aad0925e8aee8 : SUCCESS in 1h 26m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8c5c420883a437baae0cd642a57f7c3 : SUCCESS in 54m 32s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/27474403cf3f48e8885f2675476296c9 : SUCCESS in 11m 43s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/f413416c17794b86877431cd88328af9 : SUCCESS in 39m 16s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/58df85d2b83e4481b4dcdff630236a70 : SUCCESS in 36m 59s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/b0954f60f4524634b927a5255c332503 : SUCCESS in 46m 42s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/e169947f7629427d859a9cc0dc35084e : SUCCESS in 43m 09s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/6d16f2eaaafa49ccb1f90baaec3bbce4 : SUCCESS in 1h 01m 44s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/796c57043b3b4596af37c098bff08400 : SUCCESS in 1h 00m 08s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/7cef1eb4292246ab853d3ab8323cac58 : SUCCESS in 1h 10m 03s (non-voting)\n- zaqar-rally https://zuul.opendev.org/t/openstack/build/0b5c456676e54859a3dfe4ffec2899be : FAILURE in 42m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd1ff6dbcbfc887a53bd56ad1a207b54da790ad6","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2022-09-16 07:03:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"952d05929733cdd15d7900aefb7bc4c5b58aaac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 07:03:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b64b29971fb714925628f98f2f75ec0cdc14b20c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 08:44:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aa66931fd1b498792e23a53cd5f2e24\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0377c1056408494aac84e27f8ef2abbd : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40b3ff6c3da44de9b2f644b3dd45b69 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd78a6c43f3f43d2a444165d240dc10e : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34555bb7fc6848a1b17a7dff433e402e : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47258eb28d2741088f40c8824202feb2 : SUCCESS in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/4667818c2eb441f29fa8d79c9e7ccdd2 : SUCCESS in 1h 08m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f8f6d4fe7904452b1a0bf0334de8508 : SUCCESS in 1h 34m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5498d21c41d846f993f95c6e234b61a9 : SUCCESS in 1h 23m 53s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/b2333791be024c1eaa318d94b69dddea : SUCCESS in 11m 59s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/96803f9eb78e4211a73dce76a9b1e9c0 : SUCCESS in 45m 40s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/617f4429ea4b4cb5a5f9b4c9dce49592 : SUCCESS in 42m 20s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/71173bf18f374ed89634a9ec79b7deac : SUCCESS in 45m 30s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/2606db992fb24bdab2ea8a6b43ffdfb8 : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0d2955fcc63d0026fdb6d95ab4047b2803c55347","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 08:44:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"538b4ae18ab248c1ae2fd751f903c7b66db4a24d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 08:46:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa934b199a94afca367df69bb122fc9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f59f5f4a9c3f4129871fbd2c04493511 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2125930e2731ba8dac3caf30f1042b89a1c4fb4","revisions":{"b2125930e2731ba8dac3caf30f1042b89a1c4fb4":{"kind":"REWORK","_number":1,"created":"2022-08-15 02:07:54.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/45/853045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/45/853045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/45/853045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/45/853045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/45/853045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/45/853045/1"}}},"commit":{"parents":[{"commit":"4a035775a099e5603a062712529f0e5745f45420","subject":"remove unicode prefix from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/4a035775a099e5603a062712529f0e5745f45420"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 02:07:00.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-15 02:07:06.000000000","tz":480},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n\nChange-Id: I132e2e4f2c71217bc42b28b53a52a1be6d3b19b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/b2125930e2731ba8dac3caf30f1042b89a1c4fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/b2125930e2731ba8dac3caf30f1042b89a1c4fb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
