)]}'
{"id":"openstack%2Fzaqar~871265","triplet_id":"openstack%2Fzaqar~master~I5bb488f33d1c24298bb96e2702d2c98415db6cf4","project":"openstack/zaqar","branch":"master","topic":"optimize-plugin-shell","attention_set":{},"removed_from_attention_set":{"8846":{"account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"last_update":"2023-01-29 01:29:07.000000000","reason":"\u003cGERRIT_ACCOUNT_8846\u003e replied on the change","reason_account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}}},"hashtags":[],"change_id":"I5bb488f33d1c24298bb96e2702d2c98415db6cf4","subject":"Optimize devstack plugin","status":"MERGED","created":"2023-01-28 03:13:31.000000000","updated":"2023-01-29 04:27:26.000000000","submitted":"2023-01-29 04:26:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871265-optimize-plugin-shell","meta_rev_id":"3ee97c3d361ea3c61980dd2966fef3fc6c230ed0","_number":871265,"virtual_id_number":871265,"owner":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-29 04:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-28 09:36:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-28 09:36:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-28 03:24:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-28 04:42:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df771b9212bd5783657bf2fc32b3ce4f79fa42a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2023-01-28 03:13:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ca886891fd167878c98d007e70f68582491c3b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 03:24:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84ef70b8d7e048dda60dbe0755bd2826\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e84a6e01f2546b592262a7d434fc0d9 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc3f2f6c6c0546cc95ff3278cd2756b2 : RETRY_LIMIT in 2m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d7f10feaece70f67db7ff770f103263b95a035c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 04:42:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e1a5bee2f2e4ae59fede7637f365f37\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae7dbd6320194556b7d003de43642a2e : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe82a86faa714f508bffe02eb6ebf8cc : FAILURE in 9m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18fecab3668943a1b6b5b1728a1e65b9 : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3095584b568d43b2998bc27b3ab19d10 : SUCCESS in 3m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/03c68fea6d01420aa77c011d95eade07 : SUCCESS in 56m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/95471f367ef44b5e858afa6c057aa4d2 : SUCCESS in 59m 46s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0350af1cbd9240e1a3494fd097b827ca : SUCCESS in 1h 23m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77a3cd5fca2349fba3420b47a6c45335 : SUCCESS in 55m 54s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/5a0d01f83aa443b7ab45d4a0d1223f80 : SUCCESS in 11m 46s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/3ad6d347c84c4ae6bc2306b98f9bbb1c : FAILURE in 15m 43s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/4986b407e80c4a9c864c6bb188df4a16 : FAILURE in 17m 19s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/11239ca4d9a14363a150ccf5e3e52f75 : FAILURE in 15m 20s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/043f590be2304045838b84e7ab252aa8 : FAILURE in 15m 46s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/6eb9fb1cae4c402ba97af8e227786ee4 : FAILURE in 49m 06s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/8875f6ad6cbd4d50bff72e651d895422 : FAILURE in 42m 18s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/5c4cd49e5cce46709dd4322502ecb38d : FAILURE in 50m 28s (non-voting)\n- zaqar-rally https://zuul.opendev.org/t/openstack/build/13f454778a9f493bbaf89be28307e556 : FAILURE in 18m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14f2112e1b06c0be3b8da138bd664ad4020bb7a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2023-01-28 06:18:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d1d5834a3fa1ea500e9a900f01f1fe107b877f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 06:29:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/739eee30bd7e4024acda97071753104d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d68337362a23456eab914147921a1bbe : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83acccdf17ab4404a9a63c0762a294fd : RETRY_LIMIT in 2m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"25a5db715d1a7407cd76dad513bad2734c6205c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 08:00:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e564986cec45679969f2a31a5aed47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0b67b1acdd4e899747d64ade2e3007 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22aa19dd583d48ffb45b7f1967fd3168 : SUCCESS in 11m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15111a70f61340a0b9c2800096eb5263 : SUCCESS in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0246bccfe6804de78c3217a8b630a142 : SUCCESS in 4m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/3fd77814c47c4cfbab779d6c5a156f0d : SUCCESS in 1h 21m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c6156ec839ad4e0db5d921f33dcab656 : SUCCESS in 1h 08m 55s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0599dc49078142aa943668c6d3ac3062 : SUCCESS in 1h 33m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39ccfaa5e2434603a19940869f3cb748 : SUCCESS in 1h 07m 48s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/76d33a7a8a214ecb975a6e3319839f4e : SUCCESS in 11m 44s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/15c81beccd7a479fb553515b0c114b1a : SUCCESS in 54m 21s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/8f2f47e648a04abcb08f0e93f9e7b4a5 : SUCCESS in 38m 55s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/10c4016c31364aeb8c11f9cb93a00e3a : SUCCESS in 42m 13s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/ebe849bee3eb4ff29c5e7554fad8986b : SUCCESS in 53m 19s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/f7c9d631ebf140baa6a9d4f6f9b5f2e3 : SUCCESS in 1h 00m 31s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/d5224410e795474fb941c59b58e23a85 : SUCCESS in 1h 06m 06s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/085c0dc979aa4269b8b38f2c43a5e162 : SUCCESS in 1h 07m 37s (non-voting)\n- zaqar-rally https://zuul.opendev.org/t/openstack/build/02fffce0262243c3a48b5662156dcd80 : FAILURE in 45m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6beadc47cdaedb30587c30bba8d949bc04e9234","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2023-01-28 09:36:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ea9832269f96ced796deee846b936e49381270a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 09:37:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d14635ec0b002c5fc49a9faa121772d6bb45d49e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 11:33:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2238f14070c14813b6102fc18640012d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80826e8b09f74e76bad6866ac1b8e42f : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3af54efd44ff4dfe910dd913f8bcb901 : SUCCESS in 9m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/025cc2b674ff4b00ab41815f1f31b40c : TIMED_OUT in 40m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f5e2c82e74e49a9b84e8f53a995e7d8 : SUCCESS in 4m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/ec2f1b002aa9425b99ab42a0e021c265 : SUCCESS in 1h 06m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab884c6644244c59a5b88d0ea5da569e : SUCCESS in 1h 31m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ac961db3c0e4fc6b22c2d8329f1444c : SUCCESS in 1h 02m 08s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/37c63476e5f9437583b34a424427d01a : SUCCESS in 11m 47s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/fb290cc635324b1996aa617302194679 : SUCCESS in 22m 10s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/6f023893ad734ba7acffcbbc040edc09 : SUCCESS in 21m 49s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/6b09cc219d7a461cad95424547924dd7 : SUCCESS in 44m 14s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/2665e0b3943241a0973c9c0d46435d69 : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6c78353912323b855e0cb5bbf0ef20e5a2df6b31","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2023-01-29 01:29:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6cba6c48f8b7193f87d4dae1ceb00a6bfc3f895","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 01:45:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7cc6dbb76724d5a9f996d8630a74e40\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff82754abf2749b0aca57adb5e6e50f3 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/84b15e66f7ac4aea97ec2ce53a48ee9d : RETRY_LIMIT in 2m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"75e786a038bca2d0131e5fdb8f625eafdceb36f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 02:44:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee21175de34b4619bb282d9c785966e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2a8ffbb6c4c4314b60236d3a5ea59ca : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5a5fea951e541ed9addcee296901e1b : SUCCESS in 9m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/153093a3bdc5476f8c5be45a4ec96ec0 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ed2b211390461d908bde6f1859dc20 : SUCCESS in 3m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/9b229b74b2124271a97ef388d857b3b9 : SUCCESS in 42m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/802a00a2dbc54fa5b2c861bf575250f6 : SUCCESS in 42m 08s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f0c948a965af4c7ea19a61d4ae8a5fda : SUCCESS in 1h 13m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82175333641b436abea0f3c3bdd06049 : SUCCESS in 1h 03m 06s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/9d7fc030742b49f28d1b5b056e4e46d6 : SUCCESS in 11m 45s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/9603d0aa071748c4befce1c5e750e957 : SUCCESS in 21m 34s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/fa972ebf586c4dad9411882b4dc35da1 : SUCCESS in 23m 17s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/22c9072c69a540a8b4abc4d108839087 : SUCCESS in 24m 48s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/ad946b74ba324a959fc7b404c1492846 : SUCCESS in 41m 23s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/8649593cac1948599460293ebbf71ebd : SUCCESS in 43m 10s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/741007b559434721bbcc14afda6a21e3 : SUCCESS in 38m 29s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/5018b1bab03c418aac55b936b70753fe : SUCCESS in 58m 58s (non-voting)\n- zaqar-rally https://zuul.opendev.org/t/openstack/build/3b5e8fed1214458bae9eb5e4543c13ba : FAILURE in 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fa45e2e88a548e7361b10507861b681dbeb35ee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 02:45:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5db9cacb78b564803125ad91af47cba663d40dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 04:26:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c14e8560f2149deb315ca85841c6bc8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/025ebb4ae23342e8af4762be5a1bf341 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c2d9b542fbb48d899c416240870b8f8 : SUCCESS in 9m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdf42edd6cfd4a89986bbb2d1ebcb4be : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c26c759934c4ad788634cc6ca6761a0 : SUCCESS in 4m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/4f5c78473b774e4193e9251eab73c7dc : SUCCESS in 50m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6161785502f8479e92257bf3372eee1d : SUCCESS in 1h 25m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7ada9302e1b43bf96852767e610aa8d : SUCCESS in 55m 51s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/dbecfd0da7b044e6a9efd59659f48e99 : SUCCESS in 11m 54s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/6696681534984a4ba8e957e185cf746c : SUCCESS in 38m 38s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/5dadfde2a6074732bbd75b529a1497a1 : SUCCESS in 38m 44s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/992883439a44486693afbb67458273b1 : SUCCESS in 33m 16s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/a2496b515c6a4e9c9b821e5279614ec6 : SUCCESS in 30m 22s","accounts_in_message":[],"_revision_number":2},{"id":"584a05945313b9fa6f9e23f845e108576cf62e0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 04:26:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3ee97c3d361ea3c61980dd2966fef3fc6c230ed0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 04:27:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ad5d5653db04a0193fd8470109e7a48\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fa31e9b9c64ee8a995a690e72d922a : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b34431779243cd274c9d4a71bb544cae980ef52","revisions":{"296cb3b561698305cc5b2e7c88eace9e710b8c29":{"kind":"REWORK","_number":1,"created":"2023-01-28 03:13:31.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/65/871265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/65/871265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/65/871265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/65/871265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/65/871265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/65/871265/1"}}},"commit":{"parents":[{"commit":"8049697fc962c61d56f4e7ed1208fdd40dd3e0ec","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/8049697fc962c61d56f4e7ed1208fdd40dd3e0ec"}]}],"author":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2023-01-28 03:10:25.000000000","tz":480},"committer":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2023-01-28 03:10:25.000000000","tz":480},"subject":"Optimize devstack plugin","message":"Optimize devstack plugin\n\nFor fixing the issue that mongodb in not in ubuntu 22.04 release, we need\nto give a work-around way to install mongodb successfully in 22.04.\n\nChange-Id: I5bb488f33d1c24298bb96e2702d2c98415db6cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/296cb3b561698305cc5b2e7c88eace9e710b8c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/296cb3b561698305cc5b2e7c88eace9e710b8c29"}]},"branch":"refs/heads/master"},"6b34431779243cd274c9d4a71bb544cae980ef52":{"kind":"REWORK","_number":2,"created":"2023-01-28 06:18:36.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/65/871265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/65/871265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/65/871265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/65/871265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/65/871265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/65/871265/2"}}},"commit":{"parents":[{"commit":"8049697fc962c61d56f4e7ed1208fdd40dd3e0ec","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/8049697fc962c61d56f4e7ed1208fdd40dd3e0ec"}]}],"author":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2023-01-28 03:10:25.000000000","tz":480},"committer":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2023-01-28 06:18:15.000000000","tz":480},"subject":"Optimize devstack plugin","message":"Optimize devstack plugin\n\nFor fixing the issue that mongodb in not in ubuntu 22.04 release, we need\nto give a work-around way to install mongodb successfully in 22.04.\n\nChange-Id: I5bb488f33d1c24298bb96e2702d2c98415db6cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/6b34431779243cd274c9d4a71bb544cae980ef52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/6b34431779243cd274c9d4a71bb544cae980ef52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
