)]}'
{"id":"openstack%2Fzaqar~905839","triplet_id":"openstack%2Fzaqar~stable%2Fwallaby~I480ce9304a51fb2fa1818a0f6ea36116230ff506","project":"openstack/zaqar","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-18 21:23:42.000000000","reason":"Change was submitted"},"8846":{"account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"last_update":"2024-01-18 20:24:40.000000000","reason":"\u003cGERRIT_ACCOUNT_8846\u003e replied on the change","reason_account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}}},"hashtags":[],"change_id":"I480ce9304a51fb2fa1818a0f6ea36116230ff506","subject":"Remove lower-constraints job/tripleo job","status":"MERGED","created":"2024-01-17 09:05:16.000000000","updated":"2024-01-18 21:28:19.000000000","submitted":"2024-01-18 21:23:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905839","meta_rev_id":"b73d484c01629b7aa153b61d3117a24e0da8d21d","_number":905839,"virtual_id_number":905839,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 21:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-18 20:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-18 20:24:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 10:23:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-18 16:18:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"state":"REVIEWER"}],"messages":[{"id":"1438ed035086148c5029924eb343aa6bf91937a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-17 09:05:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f45d17c05c72b15fda7fac658052d5f0d08111d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 10:23:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b9e28308d9f4eabbdc43b94ca3df683\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e2664eb07c2f43b29fdfde149b089bbf : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d16b7039ecd464c8fd278fdc6adac41 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4607b029aab6442ba9194af1601e06f7 : SUCCESS in 11m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e34d3aaed6894280a3fcd37140a231de : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d00a1e957ed47859c7f7df9a72e223a : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f353add825694e088033c36059eb144b : SUCCESS in 5m 18s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/a3b12ba900f34bbf97e49f34144d872c : SUCCESS in 11m 34s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/7546871e544849b9b11040a3918ca4df : SUCCESS in 50m 58s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/99927d58827b4503a61183c2ab19ace5 : SUCCESS in 46m 04s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/774f757668b942eaa16854327431b2cd : SUCCESS in 37m 16s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/bbbf7ee3af26424f80e27029d90365ca : SUCCESS in 54m 45s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/aaca66b0453a4ff2b9ded46c8888e180 : SUCCESS in 1h 11m 36s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/280ed88572204cdfaf3b17435d776f1a : SUCCESS in 1h 09m 46s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/0fda963ae7d049f781cff8577db72f0a : SUCCESS in 1h 08m 21s (non-voting)\n- zaqar-rally https://zuul.opendev.org/t/openstack/build/85e82091f3b6495b83789fd9f660e5c6 : SUCCESS in 44m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8920414828b858b430b65cfeca49d04fdf9316b0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 16:21:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d81d01aa4c8d734515f9e65aa6b788e042e3611","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2024-01-18 20:24:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f806ebae1ec9f226fe5bfa1faa81ba9f7d49f886","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 20:25:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3de02804886c5a92397429f3cba96a3a61663c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 21:23:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a47707280714332a0987ca793ae58aa\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f8f7a94b4ff3440b98ac0621e7694cd2 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb2c848c5a04f41a80ce3540ed54b6e : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df184c2f67bc496e95910004b9b87b17 : SUCCESS in 10m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65178b993bbf408c947435c5891f04fe : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9761959bdbb4e7f8ec2e7243a857d42 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33fa5d6013044886a704760f43a90adc : SUCCESS in 5m 42s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/4bd97978371040d7a80963587fc765ae : SUCCESS in 13m 21s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/5db70f0b01974caeab7d20361606d85a : SUCCESS in 33m 10s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/492a712244aa44e2bb8951fc3e559697 : SUCCESS in 44m 17s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/8cce4457f0d646d6bdda07f3b25aa357 : SUCCESS in 37m 35s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/4eb2b1f99a314fa1928be7af9711b828 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7801e91be8df651da140c2be79bf4521d0d654bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 21:23:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b73d484c01629b7aa153b61d3117a24e0da8d21d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 21:28:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd24ffb2b8174795a01ac058ebc11097\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecda59163b4d41899339bd017134fd52 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82305c6af71441ada1305074bd511334 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1164cbdaa68e040221b0acb930ab851381cd2aeb","revisions":{"1164cbdaa68e040221b0acb930ab851381cd2aeb":{"kind":"REWORK","_number":1,"created":"2024-01-17 09:05:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/905839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/39/905839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/905839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/905839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/905839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/39/905839/1"}}},"commit":{"parents":[{"commit":"a28fa8ba24a99009e0b16eeddb818f845825168e","subject":"Update deprecated zuul syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/a28fa8ba24a99009e0b16eeddb818f845825168e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-12 04:26:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-17 09:04:21.000000000","tz":540},"subject":"Remove lower-constraints job/tripleo job","message":"Remove lower-constraints job/tripleo job\n\nDuring the last PTG, TC agreed to remove the lower-constraints job[1].\nBecause the job is currently broken, this change removes the job\naccording to the decision.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2022-April/028104.html\n\nThis change also removes TripleO job because TripleO no longer uses\nZaqar.\n\nConflicts:\n\t.zuul.yaml\n\tlower-constraints.txt\n\nChange-Id: I480ce9304a51fb2fa1818a0f6ea36116230ff506\n(cherry picked from commit 25146d215cdf3dfcb5f6c824f2e9bee40e10386e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/1164cbdaa68e040221b0acb930ab851381cd2aeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/1164cbdaa68e040221b0acb930ab851381cd2aeb"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
