)]}'
{"id":"openstack%2Fzaqar~941813","triplet_id":"openstack%2Fzaqar~master~Idec1b5861ed0eaf1882c259a61a1ccbffbdb0bf6","project":"openstack/zaqar","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-15 22:02:05.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Idec1b5861ed0eaf1882c259a61a1ccbffbdb0bf6","subject":"Replace pytz","status":"MERGED","created":"2025-02-15 14:31:06.000000000","updated":"2025-02-16 04:06:55.000000000","submitted":"2025-02-16 04:05:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941813","meta_rev_id":"1a52161433e005e18d4459d0e27d1c424ae96943","_number":941813,"virtual_id_number":941813,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-16 04:05:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-02-16 03:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-15 22:02:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-02-16 03:25:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-15 14:31:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-15 15:43:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-15 22:01:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-02-15 22:02:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"0910392269ce9a97d1c008601d0e4350629b567d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-15 14:31:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f53cda11fd7c5061db89665191fafaed22e65a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 14:31:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01deb8af63934807bca1288123c3836b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1240da141c2f4154a4563c95e49995cd : NODE_FAILURE Node(set) request 300-0026348073 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d671c3f4c1b54980b43345dac818f1ea : NODE_FAILURE Node(set) request 300-0026348074 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1681e6942260eae24058e5cd2a69065e119c868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 15:43:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d0dfb4242e442d6bf7af2b4e64b9556\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f0fe8368f0a46c3a7582568c334de83 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fba2c11927044d028b3177b4be566a1f : SUCCESS in 11m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b607103130094b1cafa5439eaf4e9adb : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc9b39faaf4e43cd9400e59302d57fea : SUCCESS in 3m 28s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/30eed8114bf7442d9c445b3a50c68d86 : SUCCESS in 9m 57s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/86e5b511cb374125948984283f4f52d3 : SUCCESS in 36m 32s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/cfe189b1414e41fdb39a2d152f390c0d : FAILURE in 24m 53s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/1f996606f2bc4f7da62b98096bd359a0 : SUCCESS in 36m 15s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/d8f3b362b1a44e73a34d97397c881065 : SUCCESS in 25m 08s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/ee360d22704245cdb11f379f02334c95 : SUCCESS in 37m 45s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/d8ea325a6b94400db65917d7a8c50a7b : FAILURE in 37m 37s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/49822b28109d47e18db149532da8a6eb : FAILURE in 1h 05m 37s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/4fbf76ab6df8480d879da05330539c0d : FAILURE in 42m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f773cdca12d0e99ebaf862cdfcbe4da8e1fe3414","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-15 15:45:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7f03628849f38a94580ca6d9d706a25802a8515","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 15:45:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51c3957380784df79533e17d21ae657f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c36901d5c88a4db8bd484406c33301d1 : NODE_FAILURE Node(set) request 300-0026348386 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5a1fc9b92b047ef8974395984d77292 : NODE_FAILURE Node(set) request 300-0026348387 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3d97c5c4126be5d12c7543e7222655fa99a06e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 16:55:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3434c99581d4b33a3f35d04b657a335\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df0e12ea5c27410bad7243ce545436f8 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/babaeb7ef05a4b35a91db58ab634b751 : SUCCESS in 10m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/812af363ffed44d48ab9ef4947cbe5d3 : FAILURE in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b647e402e220430aa4f3135c75cb1d4b : SUCCESS in 4m 19s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/55ac6911367447b38488e6219e9c8d6d : SUCCESS in 10m 21s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/eb1ef3d77a8d44a2ace5a4237899ea3d : SUCCESS in 36m 20s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/c665b7af323846fa89644872c907a6b6 : SUCCESS in 29m 15s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/4b2d6677e86d4e898eef3b96e966b84c : SUCCESS in 19m 31s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/4498d8ac06ce4577a699cfd974ad5c1e : FAILURE in 23m 19s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/53baf55367f045d2a5fa4a5e8f1d2b63 : SUCCESS in 32m 33s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/fa4196f5d1a044c1b2c3bc3b2ad5959a : FAILURE in 29m 58s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/4988c9c8e6fb43ccad5ca59b7753b8d2 : FAILURE in 37m 18s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/98b97f50c02648958facb0b56ee238d8 : FAILURE in 1h 03m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"415bbdc12ba4726e91d92317eb1561322cd0e948","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 22:01:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d98fdafafa9279c059900bbd954a5327dd82cfa3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 22:02:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"90658a3225e291e5cc39bbf55b195ad46ea51542","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 22:02:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84e756d506d54299b23680a3e26c59d7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19b1cfe0f0024f4483f63b0310a110a2 : NODE_FAILURE Node(set) request 300-0026349142 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/86eeeed2b90f442fbf14794f6d83a762 : NODE_FAILURE Node(set) request 300-0026349143 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77cbc6198e9eb229046e6a2da1d71cfbbc4cc736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 22:34:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a71c34e046c247adbeaf34d3bab98652\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5562c1ee5d7942c8bf742f5995e351a3 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f37c2b1c98214c99a35bfdb8c87a1cfe : SUCCESS in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec6fedbb0c01488f8cbbe2916faf6386 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ddebb55791d4dc284339d56459f082e : SUCCESS in 3m 39s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/7c974a6aa62d49db905d39027dcb3afb : SUCCESS in 9m 42s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/2e539fc1edea4cb187598ecf37664756 : SUCCESS in 21m 20s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/2299d0f818094d718455c61ee3958a0a : SUCCESS in 20m 15s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/397b9ceb67a549deabec93b3bdc3b500 : SUCCESS in 21m 15s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/d49e793379254bf489a66e37effcdd4b : SUCCESS in 22m 26s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/b2fac8b7922840249c60a07f6c9de8ee : SUCCESS in 23m 13s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/ae5c22ff8f9144b5b2f9dd5155e46bd3 : FAILURE in 29m 01s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/ecc1b5cfc5d0410e8df5bce2411e1790 : FAILURE in 29m 00s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/608cba3caeca471493a554eb1dab8bef : FAILURE in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"035343b7de44cd6887922906179b98dbf40bdf2d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-16 03:25:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a8f100e6293483fd784adfa18b0d8e5b79e9ac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 03:25:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25aa71b902970744b7199a62602afe87a2314174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 04:05:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f7f9a3d8f274505949dd3f7fc5f105f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa8953b48bf6468aa71fdd40b195b0b4 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0e310bd01264ae59f4108a36db4f9e7 : SUCCESS in 11m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ab83fb4a3f94877b22428c2c666584b : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cdf24f4b40145df8efe90efc3726f4a : SUCCESS in 2m 41s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/f1d86b3725b14b94bfc69eadf48e0ece : SUCCESS in 9m 34s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/aa1bbe6ce25a42a0a1f3bd4ab6d89f25 : SUCCESS in 22m 35s\n- zaqar-tempest-plugin-mongodb-jammy https://zuul.opendev.org/t/openstack/build/96c7e7545b30412fba098d4fb098a506 : SUCCESS in 22m 23s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/b7e411f154324971ae50000edaf2283f : SUCCESS in 35m 23s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/ddc39fc3e3ed4654b5d57f50852de1a6 : SUCCESS in 21m 12s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/f8542d19812640598879b45b4138e8ef : SUCCESS in 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b6f59f6eae6fcfa03fb292154ad371032a34958b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 04:05:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1a52161433e005e18d4459d0e27d1c424ae96943","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 04:06:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/226ac576296848808705154b5267000e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1787395d655843769152b73e0a41ea15 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e877f17cd38b09c568cf6b7238e7c4675562ebd9","revisions":{"e877f17cd38b09c568cf6b7238e7c4675562ebd9":{"kind":"REWORK","_number":1,"created":"2025-02-15 14:31:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/941813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/13/941813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/13/941813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/13/941813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/13/941813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/13/941813/1"}}},"commit":{"parents":[{"commit":"c224256eb559b6b4543cbd0452b23d95782e5a98","subject":"Merge \"[signed_url] secret_key should be secret\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/c224256eb559b6b4543cbd0452b23d95782e5a98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-15 14:29:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-15 14:30:59.000000000","tz":540},"subject":"Replace pytz","message":"Replace pytz\n\npytz hasn\u0027t been listed in requirements but has been installed\nexplicitly via dependency.\n\nBecause its usage can be quite easily replaced by the built-in\ndatetime module, remove it to reduce external dependencies.\n\nChange-Id: Idec1b5861ed0eaf1882c259a61a1ccbffbdb0bf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/e877f17cd38b09c568cf6b7238e7c4675562ebd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/e877f17cd38b09c568cf6b7238e7c4675562ebd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
