)]}'
{"id":"openstack%2Fzaqar~952849","triplet_id":"openstack%2Fzaqar~master~I7aae3a1c43a64b55605c88586f5a68eb0c32a211","project":"openstack/zaqar","branch":"master","topic":"devstack","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-02 16:23:38.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7aae3a1c43a64b55605c88586f5a68eb0c32a211","subject":"devstack: Remove workaround for ancient Fedora","status":"NEW","created":"2025-06-18 16:18:08.000000000","updated":"2025-07-02 16:23:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c54806d64673c3a22b5efc57ae178205ba9d6331","_number":952849,"virtual_id_number":952849,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-18 17:18:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-18 17:18:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-02 16:23:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"d8e2b9d48945b57a9311f1d167244d30b13afb44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-18 16:18:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b083ef91d175f5baef09af4e2f7ad45282c425fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 17:18:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efb98af4d5434dd688c679f324303db8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f42856f555404097d4620b3e210c1f : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86a5dc5fca414e9e87b532b89b44e813 : SUCCESS in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cad194755b2a431a819db5c1255dc663 : SUCCESS in 12m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43edeec9a4f94934acabf4cf8f8edd92 : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/267dff46168d486ebe11ebf139f7c7bd : SUCCESS in 3m 49s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/d0893de8463048f29403cc2422ec7d05 : SUCCESS in 11m 35s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/44aff59c0e824867906acabb7b3e5593 : SUCCESS in 32m 45s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/1fc5d876764744daa7f75dc5666e2976 : SUCCESS in 32m 18s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/647242a84f4f45fc929b3d7af874c702 : SUCCESS in 27m 37s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/5d422675384e447492079ec0c1ab1acd : SUCCESS in 37m 04s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/d43952f6aa704bd9abd04d268a871236 : FAILURE in 48m 18s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/551fbd0efe264ffdb533115c711a0419 : SUCCESS in 53m 49s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/cda95dc0ca804edfabc61a7a64b49ffc : SUCCESS in 52m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23135280a58112b1ab4b61ce6199a967e7fdcde4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 17:29:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c067ce9f4c564b1e9b84102adecf417a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9c62ec7bb2645e6b0ee068ada10035f : TIMED_OUT in 1h 00m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/85bea00c63824e2b8a764212e18829d3 : SUCCESS in 16m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7486d2d6a4a248a482fa20fe45d7833e : TIMED_OUT in 1h 00m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c54806d64673c3a22b5efc57ae178205ba9d6331","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-02 16:23:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e974a030d4e66fb3d3797839bd2f5dae4c27ae75","revisions":{"e974a030d4e66fb3d3797839bd2f5dae4c27ae75":{"kind":"REWORK","_number":1,"created":"2025-06-18 16:18:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/49/952849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/49/952849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/49/952849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/49/952849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/49/952849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/49/952849/1"}}},"commit":{"parents":[{"commit":"875041f4a92741aa56e508cc3ef2f36808e95e2b","subject":"Merge \"api: Resolve deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/875041f4a92741aa56e508cc3ef2f36808e95e2b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-18 16:18:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-18 16:18:00.000000000","tz":60},"subject":"devstack: Remove workaround for ancient Fedora","message":"devstack: Remove workaround for ancient Fedora\n\nChange-Id: I7aae3a1c43a64b55605c88586f5a68eb0c32a211\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/e974a030d4e66fb3d3797839bd2f5dae4c27ae75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/e974a030d4e66fb3d3797839bd2f5dae4c27ae75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
