)]}'
{"id":"openstack%2Fzaqar~962939","triplet_id":"openstack%2Fzaqar~master~I424685fd11a3df4beb0fd5292e23cab214a4f60a","project":"openstack/zaqar","branch":"master","hashtags":[],"change_id":"I424685fd11a3df4beb0fd5292e23cab214a4f60a","subject":"Install coverage only when needed","status":"NEW","created":"2025-10-03 14:37:16.000000000","updated":"2025-10-03 15:40:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cd6fb649aab238616c6ede0680766c24ff1d2431","_number":962939,"virtual_id_number":962939,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-03 14:55:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-03 14:55:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2e9106d99973b26702235a3e2bd8a0f34f7a138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-03 14:37:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a99210bd3cb6971ebc0b1572fbb8677e7ac23ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:55:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd3b069ebc664ee0b11c660d039a0520\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e54e24331a994cc88ff525ece3b8795c : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa88d95bdf194e2e873bf3ba9789a02b : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/141d7d552d5c437fb9c21cc89c6c4d96 : SUCCESS in 12m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec26acf046a940ef9bbca5d4c0e4b2f4 : SUCCESS in 12m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b7b53ee37c34cd2b90f0a06661a055a : SUCCESS in 14m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e901670436a6407980145fb67b9ce99a : SUCCESS in 4m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f3a9cb6391cc47b38d05f1d98fa86d2c : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70a22a9fc2a143fe817dceb412f39258 : SUCCESS in 4m 43s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/9a46fb6345ec4902a29e48699a38e126 : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":1},{"id":"cd6fb649aab238616c6ede0680766c24ff1d2431","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 15:40:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8203f79d223646a1893208c5b26b2331\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54adbeb0fa4a4b4aa21ad3c77a0938f6 : TIMED_OUT in 1h 00m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/70b05dd818a043bbafb2633887bf3750 : SUCCESS in 16m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ea3e0c3bc0b94ffeacf1768bacb0c830 : SUCCESS in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"888b0b2a4fd4a82c8a4ead02227cb3e63272cd48","revisions":{"888b0b2a4fd4a82c8a4ead02227cb3e63272cd48":{"kind":"REWORK","_number":1,"created":"2025-10-03 14:37:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/962939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/39/962939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/962939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/962939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/962939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/39/962939/1"}}},"commit":{"parents":[{"commit":"245ab78e066f4224f54b477143651b20eab8d3e7","subject":"Add X-OpenStack-Request-Id header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/245ab78e066f4224f54b477143651b20eab8d3e7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-03 14:36:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-03 14:37:08.000000000","tz":540},"subject":"Install coverage only when needed","message":"Install coverage only when needed\n\nThe coverage library is used only in the \"cover\" target.\n\nChange-Id: I424685fd11a3df4beb0fd5292e23cab214a4f60a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/888b0b2a4fd4a82c8a4ead02227cb3e63272cd48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/888b0b2a4fd4a82c8a4ead02227cb3e63272cd48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
