)]}'
{"id":"openstack%2Fzaqar~963039","triplet_id":"openstack%2Fzaqar~master~I46f2894ca05517c4f12bd38a513672686d80ca4a","project":"openstack/zaqar","branch":"master","hashtags":[],"change_id":"I46f2894ca05517c4f12bd38a513672686d80ca4a","subject":"Accept only supported values for auth_strategy","status":"NEW","created":"2025-10-04 12:54:08.000000000","updated":"2025-10-04 14:47:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ef75a3fece693fa71cec0b726d0bf470fa5a0524","_number":963039,"virtual_id_number":963039,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-04 14:47:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-04 13:12:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-04 14:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffb4293890432dc690890379544eee75601a3404","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-04 12:54:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7013245634b8c9ff468f0288e65937eaa4a49aa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 13:12:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cace6555de5040c09dc0fdffe069d701\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49a9b3590e89438cbac419a1a42385ba : FAILURE in 7m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b43eda8e41549aa9bb2310223837d25 : FAILURE in 7m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/910392d5e04a49f9ae0386f3a2176fae : RETRY_LIMIT in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3548861e7081a9009fbff49e0f6367db59d74c19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-04 13:45:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bbb94869134373c17567c98bbf1a1e750206a7c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-04 13:46:09.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"eaeb1ece5c53f3c3228e109afcf312f1091f91b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 14:13:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c9a08bbf70042889a3e5936508c6b97\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e462b935a0614e48882f775de1b27ac6 : SUCCESS in 14m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eeae53e19eba4b5b8589807a5708e9c2 : RETRY_LIMIT in 4m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5cf39417679e4d6ebf90e9c30a3a08ba : SUCCESS in 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef75a3fece693fa71cec0b726d0bf470fa5a0524","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 14:47:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2f4e2a7d83c4dd1b52ad9cc0d5571d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e836b73cc984e37976fe9f77d2ad881 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efb42d2a2fd0406799c2d088376b984a : SUCCESS in 10m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86b17ccd52cf4f2bab2f6533b02dd1e5 : SUCCESS in 12m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f6186f83b6b4893a2f861029e343ade : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac23eb232944fedad885c1301b5dac4 : SUCCESS in 2m 55s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/9acd2024c4f643cca5ea508abec04658 : SUCCESS in 11m 06s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/2bdff6e390e8445ab54b41adea556a9f : SUCCESS in 18m 46s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/5e64c690a9114a1c953f7669399e0bf9 : SUCCESS in 31m 01s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/e405f0579791488b87085d73bfb8e4d6 : SUCCESS in 33m 26s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/f0fbc628ff3344ca8dce18d71e1877c0 : SUCCESS in 27m 34s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/233df6978d4b40c98a3aad1bcc648a36 : FAILURE in 33m 34s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/809b13b543104218aacd5c5628f256b9 : SUCCESS in 43m 33s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/01555f4570704be1ba3f7a7ebb79df88 : SUCCESS in 59m 34s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0fc9b8953770792d34f6de404101b165c7897150","revisions":{"362fb96169428a4db8c56721f3f897990cf12bd4":{"kind":"REWORK","_number":1,"created":"2025-10-04 12:54:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/963039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/39/963039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/963039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/963039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/963039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/39/963039/1"}}},"commit":{"parents":[{"commit":"245ab78e066f4224f54b477143651b20eab8d3e7","subject":"Add X-OpenStack-Request-Id header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/245ab78e066f4224f54b477143651b20eab8d3e7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-04 12:53:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-04 12:53:59.000000000","tz":540},"subject":"Accept only supported values for auth_strategy","message":"Accept only supported values for auth_strategy\n\nChange-Id: I46f2894ca05517c4f12bd38a513672686d80ca4a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/362fb96169428a4db8c56721f3f897990cf12bd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/362fb96169428a4db8c56721f3f897990cf12bd4"}]},"branch":"refs/heads/master"},"56aac88d5ed348a84b50a13487abd6d0008c5d6e":{"kind":"REWORK","_number":2,"created":"2025-10-04 13:45:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/963039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/39/963039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/963039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/963039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/963039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/39/963039/2"}}},"commit":{"parents":[{"commit":"245ab78e066f4224f54b477143651b20eab8d3e7","subject":"Add X-OpenStack-Request-Id header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/245ab78e066f4224f54b477143651b20eab8d3e7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-04 12:53:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-04 13:45:40.000000000","tz":0},"subject":"Accept only supported values for auth_strategy","message":"Accept only supported values for auth_strategy\n\nChange-Id: I46f2894ca05517c4f12bd38a513672686d80ca4a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/56aac88d5ed348a84b50a13487abd6d0008c5d6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/56aac88d5ed348a84b50a13487abd6d0008c5d6e"}]},"branch":"refs/heads/master"},"0fc9b8953770792d34f6de404101b165c7897150":{"kind":"REWORK","_number":3,"created":"2025-10-04 13:46:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/963039/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/39/963039/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/963039/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/963039/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/39/963039/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/39/963039/3"}}},"commit":{"parents":[{"commit":"245ab78e066f4224f54b477143651b20eab8d3e7","subject":"Add X-OpenStack-Request-Id header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/245ab78e066f4224f54b477143651b20eab8d3e7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-04 12:53:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-04 13:46:09.000000000","tz":0},"subject":"Accept only supported values for auth_strategy","message":"Accept only supported values for auth_strategy\n\nChange-Id: I46f2894ca05517c4f12bd38a513672686d80ca4a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/0fc9b8953770792d34f6de404101b165c7897150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/0fc9b8953770792d34f6de404101b165c7897150"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
