)]}'
{"id":"openstack%2Fzaqar~974593","triplet_id":"openstack%2Fzaqar~master~I35816774499bab3c2b1e36e096c108373b163c73","project":"openstack/zaqar","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-24 13:51:18.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I35816774499bab3c2b1e36e096c108373b163c73","subject":"pre-commit: Bump versions","status":"MERGED","created":"2026-01-24 12:20:06.000000000","updated":"2026-01-24 14:29:46.000000000","submitted":"2026-01-24 14:28:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974593","meta_rev_id":"eb8c8facf94e63e1311cfdf7d7853dd9ffda4083","_number":974593,"virtual_id_number":974593,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-24 14:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-01-24 13:51:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-01-24 13:51:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-24 12:34:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27515e8df905399939350e840f6e88904da6f70f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-24 12:20:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e68d9312252f5bf9b58c51108cc7da5a854d69f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 12:34:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/233fbce15e7947428a0262f0ce528388\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/976313f633f34ee9a57b9717cbe35685 : FAILURE in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11c178ec56d4ebcb40a961289f4c7da : SUCCESS in 3m 45s\n\nWarning:\n  Comments left for invalid file zaqar/common/auth.py","accounts_in_message":[],"_revision_number":1},{"id":"2f7396da882e691db27466a60206c90c1944ae75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-24 12:42:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"25b5910058277acccd9966ff221e01a0ee4a057c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 13:07:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7549c9dc528d44c6b9a29bf982621f2f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a31286fa00684c76a7c3058e3578dd09 : RETRY_LIMIT in 4m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c893d516f264b948d86f5831db5bc84 : RETRY_LIMIT in 4m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d4ea382e2f714fd488b01a009491ef67 : RETRY_LIMIT in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32b017b9523fdcfd9664d1f3393bfb1404a79622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 13:50:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f17d5b861c0f48039c52de5bc614c16e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81f6ac4af8a94f09a3ff25469ad6a150 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86cb15dc466b447b93b299408f97a2ca : SUCCESS in 10m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e85014cb76744e3f92127039796b5e19 : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/298ca7974b53493fb32e6b923fa75c2d : SUCCESS in 3m 44s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/4b897d64be34435b9a1176fe9c338380 : SUCCESS in 8m 55s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/c945a6ac068b45239b02b22d5ed4a3d0 : SUCCESS in 24m 00s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/a08262faa8c24a5194ef5b118fb4816c : SUCCESS in 31m 16s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/47197f5ced1049de8a4ab537a0fd72ba : SUCCESS in 26m 08s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/25c153827b054e40b486a91655972ff9 : SUCCESS in 33m 40s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/5e73eb9c32b84c9780d9c14dc13c405b : FAILURE in 48m 28s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/7a8229846dd5410b9109916529dd8daa : SUCCESS in 38m 16s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/0b528f171b8b4491b3f04d2eeab9c7a9 : SUCCESS in 59m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a46e9c953b0687d956af86df03d13c7068e9db3a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-24 13:51:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9ff6ba8212dbdef49aaf53e5f3a18d7e4d3aff7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 13:51:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9c78513c0e8c627892e11c6e24f0869b9e1ce6af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 14:28:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cf72d8cdcd44f418f47def7a67c8ad1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee70ace57d44ff299565de7dac0dcfe : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3585b9971c043ea8fecf526857b71d6 : SUCCESS in 13m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65665cd017914e33aced40e74ecf4e5c : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb0a42d9ada4ea88927885af30edcd2 : SUCCESS in 3m 33s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/e54da438a0354c3f8d592f963e3adaff : SUCCESS in 7m 36s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/11ccbe23dd3c4f5fb3785d44083a1f90 : SUCCESS in 15m 17s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/84ea56c52b8240f1b9f33b34019e5744 : SUCCESS in 33m 25s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/91c558d2e4c04494885f75aebe2e5fc3 : SUCCESS in 33m 46s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/b392846b46174d08b997c75e8f4b1d7f : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":2},{"id":"919ad5d784c9d5cebeb52d9cbcf57110b258882f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 14:28:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"eb8c8facf94e63e1311cfdf7d7853dd9ffda4083","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 14:29:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df0b0e84268a46edab9f10b28bdcabb1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776fb17cc29a4e0db265ae2c188b432b : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"36acd1e87c2634e64f5623f2be374ee1d80d1f3f","revisions":{"876b47e14e5c1e4ec02355bdc149f04f393f62b6":{"kind":"REWORK","_number":1,"created":"2026-01-24 12:20:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/974593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/93/974593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/93/974593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/93/974593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/93/974593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/93/974593/1"}}},"commit":{"parents":[{"commit":"c06fad601e820274353bd6b0d2cc2726f06683b7","subject":"Merge \"Replace removed testcase.TestSkipped\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/c06fad601e820274353bd6b0d2cc2726f06683b7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-24 12:19:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-24 12:19:47.000000000","tz":540},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nChange-Id: I35816774499bab3c2b1e36e096c108373b163c73\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/876b47e14e5c1e4ec02355bdc149f04f393f62b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/876b47e14e5c1e4ec02355bdc149f04f393f62b6"}]},"branch":"refs/heads/master"},"36acd1e87c2634e64f5623f2be374ee1d80d1f3f":{"kind":"REWORK","_number":2,"created":"2026-01-24 12:42:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/974593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/93/974593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/93/974593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/93/974593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/93/974593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/93/974593/2"}}},"commit":{"parents":[{"commit":"c06fad601e820274353bd6b0d2cc2726f06683b7","subject":"Merge \"Replace removed testcase.TestSkipped\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/c06fad601e820274353bd6b0d2cc2726f06683b7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-24 12:19:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-24 12:42:18.000000000","tz":540},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nChange-Id: I35816774499bab3c2b1e36e096c108373b163c73\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/36acd1e87c2634e64f5623f2be374ee1d80d1f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/36acd1e87c2634e64f5623f2be374ee1d80d1f3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
