)]}'
{"id":"openstack%2Fzaqar~989873","triplet_id":"openstack%2Fzaqar~master~I1d90d7ff2c6cb6c9e9cb71cdf651fcbfe56e5763","project":"openstack/zaqar","branch":"master","hashtags":[],"change_id":"I1d90d7ff2c6cb6c9e9cb71cdf651fcbfe56e5763","subject":"Fix broken log format string","status":"MERGED","created":"2026-05-24 03:57:44.000000000","updated":"2026-05-24 06:20:33.000000000","submitted":"2026-05-24 06:19:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989873","meta_rev_id":"bc2acaa8a53c87b42a92df8366987fb9abd1a59a","_number":989873,"virtual_id_number":989873,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-24 06:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-05-24 05:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-05-24 05:41:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-24 04:33:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-24 05:01:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"129ed5266731d9458e8aeda542ade045c5c23a0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-24 03:57:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43386ff5accbbb89cc7664e75be4a76ab64e0281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-24 03:58:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5a2976e1ac5c0360b2c066a171b97fc4b7b582d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 04:33:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa98f8c5adf46ae899dc69a63b843e2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4e9d9d9d39454536a568d85716288082 : RETRY_LIMIT in 2m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bc96f8d84b2b4a46b24cfd82b7f68c17 : RETRY_LIMIT in 2m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8eb23a0e63134cf69d2a7f4a15316392 : SUCCESS in 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8026e0165496701245431c0eca4bae5b0ddf67f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 05:01:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7767e5ea40ca48bc945baf924416f4fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f995ee238d8400fa9862d1782578ad2 : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8de14d3bc5eb4fe1a30456acd912f80b : SUCCESS in 7m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a51e8d78cd46458fb9ee09d81b2015f1 : SUCCESS in 9m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c619376de0454708a03a14cdb6864cdb : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea443d72a254db0ab1771792f8f8dc5 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae480ec6321a4cada2bd3e0db9aa4b3a : SUCCESS in 2m 43s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/e08fcf565fe143af98166f7328658ca9 : SUCCESS in 8m 48s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/1b352853800342d1bdac347bee93a845 : SUCCESS in 30m 04s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/1f47c0f2d3eb4e329fd05afe220f167c : SUCCESS in 16m 29s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/6f248a5a5c5b44009d56b2b412167dea : SUCCESS in 23m 59s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/ec2bc45904a2426094b41db2ae894d43 : SUCCESS in 32m 02s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/1cbb6a49e7f14c31b72fd564af3c8f70 : FAILURE in 20m 35s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/dfbbe423f19f46e195d3612e1f3ac22c : SUCCESS in 46m 03s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/518a09ada6864866b4d6588f52ee9d2e : SUCCESS in 58m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b4fea7698124b38737f1504af1bcf6c760e6a8c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-24 05:41:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1ce300d9db50e991f3b57826a4475e734a95f93c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 05:41:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90a5d55845ce7dc801838855cdc8177391d6c3e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 06:19:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3313b804bff4ff482a3a4e293a4262c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f10734f84e4c8292833152e5c334ad : SUCCESS in 2m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a559b89e87aa485a8b516a61b19e6fb2 : SUCCESS in 9m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ed0c7b55927478b8c38d8c68e5a4b99 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb22f6a5c4842ea8bf5dd0c55dfc249 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4014fa8f96b64539a4e1a3a374a4cfdc : SUCCESS in 2m 32s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/221588f6e457435ab02958dae610a1fc : SUCCESS in 7m 56s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/034e3634074d42769a716d79f8b1d41a : SUCCESS in 20m 46s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/39e3f6caaf9248df8b6bcc6d66b0dab8 : SUCCESS in 17m 16s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/2c0c83dcf8a84175a6cc3767e0fe7306 : SUCCESS in 23m 19s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/5e82ff36f4d94432800f2ec66a3904d2 : SUCCESS in 32m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d3e554868581e963d1d1984d0c21aaab7a7e0545","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 06:19:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bc2acaa8a53c87b42a92df8366987fb9abd1a59a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 06:20:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/718159e2b68b423884bf1969ff3d97e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8caa48010d7c45fbb6a0902acab910fa : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b30b02ba7aaff047924c002986cdc131ee8b1d25","revisions":{"1cf0b08ae954aaefb5c5056316a1e43297fb728f":{"kind":"REWORK","_number":1,"created":"2026-05-24 03:57:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/989873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/73/989873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/73/989873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/73/989873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/73/989873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/73/989873/1"}}},"commit":{"parents":[{"commit":"76739e36f1573279a9f49abbb95da4e3e76cc516","subject":"Merge \"Fix unit test failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/76739e36f1573279a9f49abbb95da4e3e76cc516"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-24 03:57:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-24 03:57:34.000000000","tz":540},"subject":"Fix broken log format string","message":"Fix broken log format string\n\nChange-Id: I1d90d7ff2c6cb6c9e9cb71cdf651fcbfe56e5763\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/1cf0b08ae954aaefb5c5056316a1e43297fb728f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/1cf0b08ae954aaefb5c5056316a1e43297fb728f"}]},"branch":"refs/heads/master"},"b30b02ba7aaff047924c002986cdc131ee8b1d25":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-24 03:58:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/989873/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/73/989873/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/73/989873/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/73/989873/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/73/989873/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/73/989873/2"}}},"commit":{"parents":[{"commit":"76739e36f1573279a9f49abbb95da4e3e76cc516","subject":"Merge \"Fix unit test failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/76739e36f1573279a9f49abbb95da4e3e76cc516"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-24 03:57:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-24 03:57:53.000000000","tz":540},"subject":"Fix broken log format string","message":"Fix broken log format string\n\nFix TypeError caused by the incorrect tuple layer.\n\nChange-Id: I1d90d7ff2c6cb6c9e9cb71cdf651fcbfe56e5763\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/b30b02ba7aaff047924c002986cdc131ee8b1d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/b30b02ba7aaff047924c002986cdc131ee8b1d25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
