)]}'
{"id":"openstack%2Fzaqar~992201","triplet_id":"openstack%2Fzaqar~master~I89a6fa87c4156aff1896aedc67080d637df62fae","project":"openstack/zaqar","branch":"master","attention_set":{},"removed_from_attention_set":{"13039":{"account":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"last_update":"2026-06-12 15:39:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I89a6fa87c4156aff1896aedc67080d637df62fae","subject":"devstack: Add signed-by to MongoDB 8.0 apt source","status":"MERGED","created":"2026-06-08 09:41:00.000000000","updated":"2026-06-16 07:59:32.000000000","submitted":"2026-06-12 15:39:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992201","meta_rev_id":"bf049e90f7135aeeff014332b6a28854a3aa5887","_number":992201,"virtual_id_number":992201,"owner":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-12 15:39:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-08 11:59:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-06-08 12:11:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-06-12 14:57:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 10:47:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 11:59:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-06-08 12:11:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2026-06-08 12:11:41.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"5d3c64a415def07a70a98af8e3b5e1b9ef16cf94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2026-06-08 09:41:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f0c10bdb894ff9da1d72d238edb2fa784dd0769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 10:47:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7e5649f4e264caca1e6955c96684b32\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af75f3d4271a4d4990ac140d24cf747d : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6aa3224ef90d4115bef301cb5a3f8ddd : SUCCESS in 10m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30d5cb0b117a4357befebf9017d5e367 : SUCCESS in 11m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f6e0ffbe516f4407b7953f41e9566610 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c64ee6d3e64f408e68368432c2aacd : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b79399c164c4ac4ba6491a405a66027 : SUCCESS in 6m 50s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/2299601ee0554ecb886060cbca3f8654 : SUCCESS in 7m 58s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/eeca55ac713e45cc9ce619c3047b9740 : FAILURE in 34m 31s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/43d15f8213bf4c5ba3aa4cbd04729320 : SUCCESS in 31m 48s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/e8c5b80a11ae498caf9aff38bf2927cc : SUCCESS in 33m 20s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/f57abae2d3e34318864e227df47d2d2f : SUCCESS in 19m 38s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/e07a1303c3fb4a94a6a08fb2f10b797a : FAILURE in 22m 10s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/6f595e9e3e4c43128a29303eb949331d : SUCCESS in 55m 27s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/29339a4638e84d6490c674314f0cbfbc : SUCCESS in 57m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"856e0e59de332dc8ab22e019a3ca6674106cbb39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 10:47:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfd786016c1f41e3b9f4c13c62fe6677\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3778ed0778074da9b18a091732a652be : RETRY_LIMIT in 3m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/439f61e74e814ca1b80a6324f258f4ff : RETRY_LIMIT in 3m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c6fd0b0c64564ac6af85e5ed68f9f144 : SUCCESS in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ddacec5bb5f0dc8ce7d57a92f63f80d2009c4609","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2026-06-08 11:23:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62dc1ca416b9edbecfb9c9ff82072d4e4250aacf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-08 11:59:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8bfcff496c53b82782bcbe54add2bd8dc5aad70b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-08 12:11:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4233675aee69d58e79a4aee62f9bcc9a2e090c4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-08 12:11:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4c31cbfaf01cf6c1217b2e65adfc08bc8edf94cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 12:50:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0521281fa5245d08aaf9bbface403d4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee62445da3bd46af8d76343626339018 : RETRY_LIMIT in 2m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/83a83924af5b4e9d954e73da4d61ad90 : RETRY_LIMIT in 3m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9d7a86fab96a41e5ba37a9b919048593 : SUCCESS in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ab19ac5f2855ac402fd41a5cfdab92b2df2d8e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 13:13:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49a1e66fb3f3426b85e7a6bf36d9969f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4178ce0e3dc14273a53335719ae3f103 : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a31f938974c444d98036cd28527c576b : SUCCESS in 10m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f28470ba3314e579d427b18b8d15646 : SUCCESS in 7m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f853fb2351dd41a4bf8dab9bbfa18934 : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3120dc34822149fb9d9eb782fd869dc7 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71e4e2861bd246698dc3fdb92e95fd62 : SUCCESS in 5m 48s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/d64d5d5679ad48cda24ac9edec09d5f2 : SUCCESS in 7m 46s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/aa82d05f445f4430a7c08d9278ed2fe9 : SUCCESS in 16m 59s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/528a87130f7b42658b9411b3c5004b52 : FAILURE in 39m 20s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/bfdd48039318439d8266af1cf200a1d9 : SUCCESS in 34m 53s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/c1202cdf7d9548e2aaed2816f4954cc8 : SUCCESS in 18m 25s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/896e061116224a898eda32d0aad690ef : FAILURE in 25m 14s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/ad7e76a104e1425880fc6bbcddf62f14 : SUCCESS in 56m 17s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/dcc38eeb024e48b1a6a6b559ced36bc4 : SUCCESS in 31m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcce18e05094e88c4a6f2373af75fdbfd963fe47","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2026-06-12 13:17:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"007ad942fbd949c99685ba846bdb67b57d854679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 14:22:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cf0b8c7d0a54995a15a07fb220a66b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/253dfb1921ce4b178789ff4a3307f96f : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b5fb7de119343b986dc1e0a88195fda : SUCCESS in 9m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a60cdd5b12e40aab905decc9675b26c : SUCCESS in 11m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1d05b0d10ce64918b15ab77341c88e52 : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d5c92568104be0a8587da42a843386 : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd0d643c80ff4928969771cc221db57d : SUCCESS in 2m 45s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/c9065320e2aa44809ab4ea3cf7c551f6 : SUCCESS in 6m 16s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/17d44338db9344a890434890b8e95a39 : SUCCESS in 25m 16s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/a2e665aba7e54bebb6b63781e1420458 : SUCCESS in 30m 28s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/a0412c6abc44483d9bafef9946167226 : SUCCESS in 33m 47s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/80460223daed4aebb2954ba9de6dc64a : SUCCESS in 31m 17s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/a85cd1250123453c8a6d90efd6aac6c5 : FAILURE in 52m 33s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/ad6bc8b994b649cabae392ea47147469 : SUCCESS in 37m 08s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/80ff9b9e6e214cdcbc3586ccb418409d : FAILURE in 57m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ed274db40224fc32aadd1b4146a95f3217e294d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-12 14:57:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff3fc96f43031a2172f9ee8edb76119c0df89716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 14:58:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0655319684cacc09f02f8b56103701a627744ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 15:39:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ebd5897a4ab4e2b9d08f2d65634eee6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3672af027e924965a4083166c2e95277 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21b7e97ecec3454e86222a55ed58f981 : SUCCESS in 7m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4fe2ebfb9f74c4ea40dabe848aac4b4 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b17eb248234e6882dec001ebf32d0a : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c58942ae707d4d86b5f0e7683297db51 : SUCCESS in 5m 04s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/dd6f9399dd684d76a9d33b4ac6bb5d59 : SUCCESS in 7m 54s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/4c0e278bf8b142bbbb88c542dbc23b22 : SUCCESS in 14m 40s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/108b24b2f1f14337af7b3d1b17d2cc4e : SUCCESS in 33m 09s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/ab2ac3245f9e42f993af4274f8f7a6dc : SUCCESS in 36m 42s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/e37f85b8649f4d7d9c397bb47e5dc735 : SUCCESS in 33m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ebf362a5403ca93a18eb119b44314f9a0bd655dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 15:39:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d37a97930bdcca0d45c9df82fdddeeb8c2a4d42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 15:41:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b99c3fafd864ca8a3e74aa753fdeea1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/274a1651c90248f2a967483fd6204e1e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1},{"id":"bf049e90f7135aeeff014332b6a28854a3aa5887","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 07:59:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc81950cf5b14a63b10e5b469afbd7e6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0b69a47ffd1d41baad4f02414251bdd0 : RETRY_LIMIT in 3m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/301e533ff36440c2bacde9f35a749ed5 : RETRY_LIMIT in 2m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/089d8f0b4e084c8baf6662509aed6420 : RETRY_LIMIT in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"679d0f6ba91a002ec23e6ab5ed8a068cc74ad072","revisions":{"679d0f6ba91a002ec23e6ab5ed8a068cc74ad072":{"kind":"REWORK","_number":1,"created":"2026-06-08 09:41:00.000000000","uploader":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"ref":"refs/changes/01/992201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/01/992201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/01/992201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/01/992201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/01/992201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/01/992201/1"}}},"commit":{"parents":[{"commit":"244df2c029eb119e75369c0348435d698e3c4bf2","subject":"Merge \"devstack: Honor global debug flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/244df2c029eb119e75369c0348435d698e3c4bf2"}]}],"author":{"name":"Martin André","email":"m.andre@redhat.com","date":"2026-06-08 09:39:23.000000000","tz":120},"committer":{"name":"Martin André","email":"m.andre@redhat.com","date":"2026-06-08 09:40:46.000000000","tz":120},"subject":"devstack: Add signed-by to MongoDB 8.0 apt source","message":"devstack: Add signed-by to MongoDB 8.0 apt source\n\nThe MongoDB 8.0 apt repository configuration for Ubuntu 24.04 downloads\nthe GPG key into /usr/share/keyrings/mongodb-server-8.0.gpg but the deb\nsource line does not reference it via signed-by. On standard Ubuntu\n24.04 systems, apt treats unsigned repositories as a hard error, causing\nstack.sh to fail with exit code 100.\n\nAdd signed-by\u003d/usr/share/keyrings/mongodb-server-8.0.gpg to the deb\nline in both devstack/plugin.sh and tools/test-setup.sh, matching the\npattern already used for Debian 12/13 in test-setup.sh.\n\nCloses-Bug: #2155845\nChange-Id: I89a6fa87c4156aff1896aedc67080d637df62fae\nSigned-off-by: Martin André \u003cm.andre@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/679d0f6ba91a002ec23e6ab5ed8a068cc74ad072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/679d0f6ba91a002ec23e6ab5ed8a068cc74ad072"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
