)]}'
{"id":"openstack%2Fzaqar~993913","triplet_id":"openstack%2Fzaqar~master~I99856434debcf71a1400a827a78286185aefe3f0","project":"openstack/zaqar","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-18 14:35:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I99856434debcf71a1400a827a78286185aefe3f0","subject":"Resctict API access to project scope","status":"NEW","created":"2026-06-18 13:22:50.000000000","updated":"2026-06-18 15:52:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2308b5b4a3273f7a57cdcdce66c48556f392db78","_number":993913,"virtual_id_number":993913,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-18 15:46:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-18 13:57:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2ea88d07e042a66e1fdc730ec3a73618dcd9434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-18 13:22:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a847438fbba2d00fe2911d01774111b08f805f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 13:57:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7c43be338cd48dcb6d3e0074b3e3116\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17a56c568beb4c64b08a21704c3b6e41 : SUCCESS in 2m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6edf7b40c7b24cb384496deaabc56671 : SUCCESS in 8m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6dd60173197c43598eab25806b84e861 : SUCCESS in 7m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71a92f1383ab48bf99577a5c1d93a4fc : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56de20b60f2b420aabbdfceb7aae6458 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/774a7b1af91840d09fefa8c79e5503c7 : SUCCESS in 3m 18s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/8b6cb8951aec4cb78002859ac7aaadf9 : SUCCESS in 6m 04s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/0b4dde8205e8447c8403c9bac5a9b75d : FAILURE in 16m 00s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/428da5dca53e4ee3a5fdd69b646fe743 : FAILURE in 15m 23s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/3cc43f1c878c4847906d25f69bd4d52a : SUCCESS in 16m 41s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/e809510ccc154286b76795ba0ebfab15 : SUCCESS in 16m 23s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/323a01428c304da59a45c7144a3b7a42 : FAILURE in 21m 17s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/0f9a2269f5bb4bf2bd72b78119a0f676 : SUCCESS in 24m 57s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/d6a1fc57f711451eb6ba3f66a9809906 : SUCCESS in 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cacc4dde0cdf2b11284fbf8e79950b9bbe974389","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 14:10:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b5414ced61f476c931880b2a4d50a3b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/279ab24f944248d395216768719071f4 : RETRY_LIMIT in 3m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6375c5984c7c4014922f3a49dfb8b49a : RETRY_LIMIT in 4m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/62657d79d71a4b76ad329ccc971d6f36 : SUCCESS in 18m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84940200ff8ce5b4d4109a8e7bcb750cc50e3f74","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-18 14:35:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfc95ff153812c222c5fab8e399f2beae851ed39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 15:46:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/785715379c494da88d3c8b9cdd001b8f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a2a1bc8a204ed9904cb3651615e5e6 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e73717e8558e4d969644f8dd4265c000 : SUCCESS in 7m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4df984703243439ba624b648c57606df : SUCCESS in 7m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/342f11a93ba84e2695454b2d10fcdf68 : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495d902d847e47478ec6ac887536feb1 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6126652d45474da4aeb6329a8a4da8a4 : SUCCESS in 4m 16s\n- zaqar-tox-integration https://zuul.opendev.org/t/openstack/build/030516b7f40347d1b7119b5783779ee0 : SUCCESS in 10m 01s\n- zaqar-tempest-plugin-mongodb https://zuul.opendev.org/t/openstack/build/84eb32364ff34b49a2fd95ea192a0ac6 : SUCCESS in 15m 17s\n- zaqar-tempest-plugin-redis https://zuul.opendev.org/t/openstack/build/668ea7e64643439fbc2b1f5a65b3a68d : SUCCESS in 17m 04s\n- zaqar-tempest-plugin-swift https://zuul.opendev.org/t/openstack/build/448ee9a960404ff28f12e3828111baa0 : SUCCESS in 39m 04s\n- zaqar-tempest-plugin-swift-ipv6 https://zuul.opendev.org/t/openstack/build/ad6b33817b1d4f9988830cfe3ded31c8 : SUCCESS in 18m 04s\n- zaqar-grenade-mongodb https://zuul.opendev.org/t/openstack/build/9a0e6b9b07f246c5bc1f23f02c23a6ef : FAILURE in 52m 35s (non-voting)\n- zaqar-grenade-redis https://zuul.opendev.org/t/openstack/build/a57f69c2b4894227843410f1763e74b5 : SUCCESS in 50m 08s (non-voting)\n- zaqar-grenade-swift https://zuul.opendev.org/t/openstack/build/51b91808c62e45b48e0a81dd6a04bdd1 : SUCCESS in 1h 05m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2308b5b4a3273f7a57cdcdce66c48556f392db78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 15:52:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71ea073e8cce41fa95bcbd4baf4eb675\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f71e87abcfbb4102879867d3258a3e18 : RETRY_LIMIT in 3m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4ca6402b416a4ffe9ff2c5516f74ae98 : RETRY_LIMIT in 4m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b2141bdecd204546bc3ed3f8df6002dc : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e4c9cfc29d23c4a309fcd91f2197dd16d74bd57","revisions":{"8e4c9cfc29d23c4a309fcd91f2197dd16d74bd57":{"kind":"REWORK","_number":1,"created":"2026-06-18 13:22:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/993913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zaqar","ref":"refs/changes/13/993913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/13/993913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/13/993913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zaqar refs/changes/13/993913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zaqar refs/changes/13/993913/1"}}},"commit":{"parents":[{"commit":"eb6d32330cfbfd52af147e7ec8ac0dc82a5e4c4c","subject":"Merge \"Use datetime.UTC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/eb6d32330cfbfd52af147e7ec8ac0dc82a5e4c4c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 13:21:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-18 13:21:46.000000000","tz":540},"subject":"Resctict API access to project scope","message":"Resctict API access to project scope\n\nAll of the API endpoints in v2 requires project scope.\n\nChange-Id: I99856434debcf71a1400a827a78286185aefe3f0\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/8e4c9cfc29d23c4a309fcd91f2197dd16d74bd57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zaqar/commit/8e4c9cfc29d23c4a309fcd91f2197dd16d74bd57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
