)]}'
{"id":"openstack%2Fzun-tempest-plugin~1002890","triplet_id":"openstack%2Fzun-tempest-plugin~master~I1eab8f987905db3e3f2a0ada968d83dab03a9cc2","project":"openstack/zun-tempest-plugin","branch":"master","attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-29 23:53:39.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-08-29 23:53:31.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-30 10:58:58.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I1eab8f987905db3e3f2a0ada968d83dab03a9cc2","subject":"Improve container state transition error handling","status":"NEW","created":"2026-08-29 21:40:41.000000000","updated":"2026-09-12 16:26:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f46539919da677aeea3b8b501eb45839a1a80f65","_number":1002890,"virtual_id_number":1002890,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-12 16:26:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-29 22:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-29 23:53:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-08-29 23:53:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"b3a054c80284d4fac7640accd63d5a025fb513ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 21:40:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e33d60e0cccf62fa755a88d47d00f9a0f8d6286","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 21:48:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"be6b4981c61e423c90bd7c4623879a6060d69c9e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 22:00:40.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"58dee38aed425032449bfc744934148caa3a8ce3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 22:02:35.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"fd34b51e71b112cd54856109a7f8a592b5ce44e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 22:51:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7433893071bd45989b644dc60d2b0115\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21e4b56bdee4977a13b9866331ab9be : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8febc7a057e04daaa593d113f0c294d3 : SUCCESS in 2m 40s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/041b3b840a91410abef347fddaef6c28 : SUCCESS in 32m 50s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/6d43c692b64b4ea2aed0b209e1bae866 : SUCCESS in 32m 52s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/077c39940ae64af09bf9854bde755f7b : SUCCESS in 27m 11s\n- zun-tempest-py3-docker-sql-2026-1 https://zuul.opendev.org/t/openstack/build/c90d19a93f684f38b960a9290adfa04d : SUCCESS in 31m 20s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d518131d3f454ab1a8d4dbc640aa54fd : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":4},{"id":"da72a1633105c52d4a7aeeb1129ba7c1325f69b5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 22:54:52.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"82d95c77140268d834c497887ab0a62f0072001b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 22:54:54.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"e202d7ca9682ba0d6a4d9780fcc25e392bf8bed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 23:42:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ccbc9ccd2684af3b263ed3e85c484d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1d054a3843e486d912a81fd960c1ce7 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c624b2b13e384d74bbd908cc43c36dd3 : SUCCESS in 4m 04s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/af833681d1844c8aae430e72ed3e7ade : SUCCESS in 41m 45s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/4487fb7a2ae44c54bf72a7d6b6475abe : SUCCESS in 31m 58s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/47bc0e21ada44edfb5d24021c570a537 : SUCCESS in 42m 10s\n- zun-tempest-py3-docker-sql-2026-1 https://zuul.opendev.org/t/openstack/build/913c3c3aab784bf6964ad4a1b067559c : SUCCESS in 23m 03s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/0bc1eebf684041cf97b4a5120302a659 : SUCCESS in 26m 40s","accounts_in_message":[],"_revision_number":5},{"id":"aab9fa28eb31fde1212102af7d59249cb6bd1694","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:47:47.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"d30a4554a8802f5e68223513dba9b33ad9b4f33b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:47:58.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"7397d2ee112f3cfb21f5708680b47b099a747e5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 02:51:16.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"2c0e3dacbf2abb6f19c2d9578bd10e4522d0c002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 04:09:42.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aca7673d867549619c813229436190ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f79de6913bc49c99f277070d9eac2ab : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de528f049db4703bb6fbd9c110b62d3 : SUCCESS in 3m 34s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b6f9efaf07c541b6a822dfe4d6397a3b : FAILURE in 36m 44s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/067c0d010b6d4409a4d9405711e493a8 : FAILURE in 37m 29s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/4ad1b3e744464d05895f2d0161377d9d : FAILURE in 34m 40s\n- zun-tempest-py3-docker-sql-2026-1 https://zuul.opendev.org/t/openstack/build/ea88f4fc4c2240419246dbdeea8bf1e8 : FAILURE in 23m 13s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/98fef3ebfb27421c86c142ab7cfd5a2d : FAILURE in 39m 22s","accounts_in_message":[],"_revision_number":8},{"id":"3f2b8d936c4a368b86eb0105f9cf34852f6a658e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 10:58:40.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"3e0c947552cd1ba93bae389e4b26827dc02d9284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 11:55:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c506190113c146bf8cfe5a220507050f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b68434dca4bc445ca59eb5dbec159ab6 : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad49a1a3f50421883c368188b767f85 : SUCCESS in 2m 18s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/58021f77ac3649f5ba764e12c3059202 : SUCCESS in 44m 42s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/540074d74f8c4f808b7a60780ae9841c : SUCCESS in 22m 58s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/f53a1256d1e44e4b9088e052cf7e6b4e : SUCCESS in 50m 51s\n- zun-tempest-py3-docker-sql-2026-1 https://zuul.opendev.org/t/openstack/build/a86d2734f0204d2989eb9d86c5bb7def : SUCCESS in 46m 36s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/84cc0afaf38f43ad8731bd5ea6892613 : SUCCESS in 31m 05s","accounts_in_message":[],"_revision_number":9},{"id":"823328484ee06c33e7a7c7015b49ae5915eca568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-12 15:35:26.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"f46539919da677aeea3b8b501eb45839a1a80f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 16:26:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d0d9e06a59e4c2086b063cdb1882c44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44abe3e4dd604bacb9b208522fcde3a1 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d119ed4685e4203a1a9dd81fd1f8611 : SUCCESS in 2m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/397b3f64a9114b7b9d885a04c6b7b90d : SUCCESS in 33m 53s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/48d0d4a890ae425ba8e821be6231b881 : SUCCESS in 49m 09s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/902b356b080644cf98f332b22c2284aa : SUCCESS in 23m 36s\n- zun-tempest-py3-docker-sql-2026-1 https://zuul.opendev.org/t/openstack/build/7e3b117aa2ef4efb889e9cf59c0cc1ad : SUCCESS in 42m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/223b7e7cb7dd4fdc9ed6a7fb7c8363be : SUCCESS in 26m 21s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"aacd8cc7380b773df48b13ac92058294232fc999","revisions":{"29c87edbc21a4bfd49f86bad4e780155eb75e89f":{"kind":"REWORK","_number":1,"created":"2026-08-29 21:40:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/1002890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/90/1002890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/1"}}},"commit":{"parents":[{"commit":"1c8845645f750337a7b9bf13a66af52562ada80a","subject":"Ensure that images for tempest tests configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I1eab8f987905db3e3f2a0ada968d83dab03a9cc2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/29c87edbc21a4bfd49f86bad4e780155eb75e89f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/29c87edbc21a4bfd49f86bad4e780155eb75e89f"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"fdbddbafcd66659be7465b12e38884dac75c73fc":{"kind":"REWORK","_number":2,"created":"2026-08-29 21:48:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/1002890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/90/1002890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/2"}}},"commit":{"parents":[{"commit":"1c8845645f750337a7b9bf13a66af52562ada80a","subject":"Ensure that images for tempest tests configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:48:47.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I1eab8f987905db3e3f2a0ada968d83dab03a9cc2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/fdbddbafcd66659be7465b12e38884dac75c73fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/fdbddbafcd66659be7465b12e38884dac75c73fc"}]},"branch":"refs/heads/master"},"a86be4f02926a0785aa4537408c00d489ae1ebdf":{"kind":"REWORK","_number":3,"created":"2026-08-29 22:00:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/1002890/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/90/1002890/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/3"}}},"commit":{"parents":[{"commit":"1c8845645f750337a7b9bf13a66af52562ada80a","subject":"Ensure that images for tempest tests configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 22:00:40.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I1eab8f987905db3e3f2a0ada968d83dab03a9cc2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a86be4f02926a0785aa4537408c00d489ae1ebdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a86be4f02926a0785aa4537408c00d489ae1ebdf"}]},"branch":"refs/heads/master"},"2b5d8aac9b28e663d373d87982f8bb440d2456ff":{"kind":"REWORK","_number":4,"created":"2026-08-29 22:02:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/1002890/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/90/1002890/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/4"}}},"commit":{"parents":[{"commit":"1c8845645f750337a7b9bf13a66af52562ada80a","subject":"Ensure that images for tempest tests configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 22:02:35.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I1eab8f987905db3e3f2a0ada968d83dab03a9cc2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/2b5d8aac9b28e663d373d87982f8bb440d2456ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/2b5d8aac9b28e663d373d87982f8bb440d2456ff"}]},"branch":"refs/heads/master"},"a298d13852e973ceb5959a24b1671f1480da1fe5":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-08-29 22:54:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/1002890/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/90/1002890/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/5"}}},"commit":{"parents":[{"commit":"1c8845645f750337a7b9bf13a66af52562ada80a","subject":"Ensure that images for tempest tests configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 22:54:52.000000000","tz":0},"subject":"Add container state wait logging","message":"Add container state wait logging\n\nChange-Id: I1eab8f987905db3e3f2a0ada968d83dab03a9cc2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a298d13852e973ceb5959a24b1671f1480da1fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a298d13852e973ceb5959a24b1671f1480da1fe5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"45e69117d509f0e0bbbe66591381c9dfc97143a5":{"kind":"REWORK","_number":6,"created":"2026-08-30 02:47:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/1002890/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/90/1002890/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/6"}}},"commit":{"parents":[{"commit":"1c8845645f750337a7b9bf13a66af52562ada80a","subject":"Ensure that images for tempest tests configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:47:46.000000000","tz":0},"subject":"Add container state wait logging","message":"Add container state wait logging\n\nChange-Id: I1eab8f987905db3e3f2a0ada968d83dab03a9cc2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/45e69117d509f0e0bbbe66591381c9dfc97143a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/45e69117d509f0e0bbbe66591381c9dfc97143a5"}]},"branch":"refs/heads/master"},"164e38a41ef0fd75b258ee5097b73698a613a034":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-08-30 02:47:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/1002890/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/90/1002890/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/7"}}},"commit":{"parents":[{"commit":"1c8845645f750337a7b9bf13a66af52562ada80a","subject":"Ensure that images for tempest tests configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:47:58.000000000","tz":0},"subject":"Improve container state transition error handling","message":"Improve container state transition error handling\n\nChange-Id: I1eab8f987905db3e3f2a0ada968d83dab03a9cc2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/164e38a41ef0fd75b258ee5097b73698a613a034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/164e38a41ef0fd75b258ee5097b73698a613a034"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9563c904afe4203de639afaae726a4293dcb7a1a":{"kind":"REWORK","_number":8,"created":"2026-08-30 02:51:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/1002890/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/90/1002890/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/8"}}},"commit":{"parents":[{"commit":"1c8845645f750337a7b9bf13a66af52562ada80a","subject":"Ensure that images for tempest tests configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 02:51:16.000000000","tz":0},"subject":"Improve container state transition error handling","message":"Improve container state transition error handling\n\nChange-Id: I1eab8f987905db3e3f2a0ada968d83dab03a9cc2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9563c904afe4203de639afaae726a4293dcb7a1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9563c904afe4203de639afaae726a4293dcb7a1a"}]},"branch":"refs/heads/master"},"de9e69b1083f31492b62afdbdffa2f0fcf904d53":{"kind":"REWORK","_number":9,"created":"2026-08-30 10:58:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/1002890/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/90/1002890/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/9"}}},"commit":{"parents":[{"commit":"1c8845645f750337a7b9bf13a66af52562ada80a","subject":"Ensure that images for tempest tests configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1c8845645f750337a7b9bf13a66af52562ada80a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 10:58:40.000000000","tz":0},"subject":"Improve container state transition error handling","message":"Improve container state transition error handling\n\nChange-Id: I1eab8f987905db3e3f2a0ada968d83dab03a9cc2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/de9e69b1083f31492b62afdbdffa2f0fcf904d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/de9e69b1083f31492b62afdbdffa2f0fcf904d53"}]},"branch":"refs/heads/master"},"aacd8cc7380b773df48b13ac92058294232fc999":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-09-12 15:35:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/1002890/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/90/1002890/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/90/1002890/10"}}},"commit":{"parents":[{"commit":"b3ceb09fd175b2a90523cef27dfa4717a328d708","subject":"Merge \"Remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b3ceb09fd175b2a90523cef27dfa4717a328d708"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 21:40:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-12 15:35:26.000000000","tz":0},"subject":"Improve container state transition error handling","message":"Improve container state transition error handling\n\nChange-Id: I1eab8f987905db3e3f2a0ada968d83dab03a9cc2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/aacd8cc7380b773df48b13ac92058294232fc999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/aacd8cc7380b773df48b13ac92058294232fc999"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"1c8845645f750337a7b9bf13a66af52562ada80a","ours":"de9e69b1083f31492b62afdbdffa2f0fcf904d53","theirs":"b3ceb09fd175b2a90523cef27dfa4717a328d708","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
