)]}'
{"id":"openstack%2Fzun-tempest-plugin~1002965","triplet_id":"openstack%2Fzun-tempest-plugin~master~I04b872f5fb837c720e458ad4f4975e1271e4cef0","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I04b872f5fb837c720e458ad4f4975e1271e4cef0","subject":"Update requirements for prevent problems with previous releases","status":"MERGED","created":"2026-08-30 11:45:10.000000000","updated":"2026-08-30 13:24:20.000000000","submitted":"2026-08-30 13:23:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002965","meta_rev_id":"47128ba35976a6d694d52cb08d6e4b1c0210e43b","_number":1002965,"virtual_id_number":1002965,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-30 13:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-30 12:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-30 12:57:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-30 12:39:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a285f87d290958c953569752940e731cfa7aab90","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 11:45:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9069a50323b21e3b9871afda9edf7afa92d04d27","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 11:46:03.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2f8b8b15c65a758f8adf2f154e778de94bea14fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 11:47:18.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fd4f4aa0b170524c7a09111a69684a00b3ab9558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 12:39:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de0274a4d95b4ec1b22fd3f657d19078\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cfd23552eb3c440ca3e8de75e07efa2a : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d353b81d599a4abd99a53932e4c9e112 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e89173e6b97d4f9f80c9914ae48fb53a : SUCCESS in 3m 01s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/a625a8a5f10c4f21bd38403693116ad3 : SUCCESS in 44m 05s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/2ca6204780354b65b9b31cc0951fbfe1 : SUCCESS in 43m 31s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/8c5ee45d479346579b9e184fc9db800f : SUCCESS in 44m 35s\n- zun-tempest-py3-docker-sql-2026-1 https://zuul.opendev.org/t/openstack/build/2fd12d06105e483bb74facf1a45f9b57 : SUCCESS in 48m 14s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/14ea7aa9e17340aeb029d1b63ac81e0a : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":3},{"id":"268efc65537cdf2a5bbb593540b8029bba77a901","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 12:57:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"989ed62f3a3c88d504da6de368cdcbd6d841aac2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 12:57:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"de2c502c41c030b99092e8c5184c3d2bdd1fad9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 13:23:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bfebc855d39488aaa9c1b9c939e1167\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/594069cd2c2e4b1b9f45f16fc24cab61 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/040ae251411a444f84f5550d90100a08 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95fbc9546f8f459c828edd60c08cd08a : SUCCESS in 4m 04s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/bdaf0d014add4394b5122f89c767d8c2 : SUCCESS in 25m 04s","accounts_in_message":[],"_revision_number":3},{"id":"4f7657d40de2613132a7eb5e279ffa0610f0197b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 13:23:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"47128ba35976a6d694d52cb08d6e4b1c0210e43b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 13:24:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7712a6d213ac4f7b86a3ceba4ee766af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5e99912f97490f8778107a0baacc73 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"390d389576eb031c5a98dea70825b62c6fe15103","revisions":{"d9224c9af4a62a4b435a7a254906f1143cc5a9d6":{"kind":"REWORK","_number":1,"created":"2026-08-30 11:45:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/1002965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/65/1002965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/1"}}},"commit":{"parents":[{"commit":"7b5cecaa4938c2bd0b8197a70ada015936fb2cd0","subject":"Merge \"Remove unused .mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/7b5cecaa4938c2bd0b8197a70ada015936fb2cd0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 11:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 11:45:10.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I04b872f5fb837c720e458ad4f4975e1271e4cef0\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d9224c9af4a62a4b435a7a254906f1143cc5a9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d9224c9af4a62a4b435a7a254906f1143cc5a9d6"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"5a02649a9b4f1eb0851a973362c2cfc17c06aab8":{"kind":"REWORK","_number":2,"created":"2026-08-30 11:46:03.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/1002965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/65/1002965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/2"}}},"commit":{"parents":[{"commit":"7b5cecaa4938c2bd0b8197a70ada015936fb2cd0","subject":"Merge \"Remove unused .mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/7b5cecaa4938c2bd0b8197a70ada015936fb2cd0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 11:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 11:46:03.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I04b872f5fb837c720e458ad4f4975e1271e4cef0\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5a02649a9b4f1eb0851a973362c2cfc17c06aab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5a02649a9b4f1eb0851a973362c2cfc17c06aab8"}]},"branch":"refs/heads/master"},"390d389576eb031c5a98dea70825b62c6fe15103":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-08-30 11:47:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/1002965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/65/1002965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/65/1002965/3"}}},"commit":{"parents":[{"commit":"7b5cecaa4938c2bd0b8197a70ada015936fb2cd0","subject":"Merge \"Remove unused .mailmap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/7b5cecaa4938c2bd0b8197a70ada015936fb2cd0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 11:45:10.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 11:47:18.000000000","tz":0},"subject":"Update requirements for prevent problems with previous releases","message":"Update requirements for prevent problems with previous releases\n\nChange-Id: I04b872f5fb837c720e458ad4f4975e1271e4cef0\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/390d389576eb031c5a98dea70825b62c6fe15103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/390d389576eb031c5a98dea70825b62c6fe15103"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
