)]}'
{"id":"openstack%2Fzun-tempest-plugin~1002967","triplet_id":"openstack%2Fzun-tempest-plugin~master~I182acc6582225c6da7278ea10c95ef956f668fd2","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I182acc6582225c6da7278ea10c95ef956f668fd2","subject":"Return support for Python 3.10","status":"MERGED","created":"2026-08-30 13:59:49.000000000","updated":"2026-08-30 15:23:57.000000000","submitted":"2026-08-30 15:22:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002967","meta_rev_id":"3ae2836993c34119f7035bbaed975bca7c0840d0","_number":1002967,"virtual_id_number":1002967,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-30 15:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-30 14:00:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-30 14:00:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-30 14:50:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1b4d6335a1f77de8b6c95b8b615be69392fd92b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 13:59:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f60565162727cf2f16c1ac152f140963467cecdd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 14:00:46.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6134a36458b25a64cbee9cefcf804deaf97979b4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 14:00:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"01f815a890ed9a84918594fef63f941a7b28e011","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 14:00:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"53cce788615254abd3b4bd9d42f9765623246c55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 14:50:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0805c0e6d39d4a88a965b94e4ae7413c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dcc3d418dcdc4c70b32b1f18c5352579 : SUCCESS in 2m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aae4b7d1978e47d0832b0881f724b597 : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/456616b44d6f4f47bdc1ff3a449c8fd8 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c5f4a8d5ef41eabe6391e0d4b5a2d9 : SUCCESS in 1m 51s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/be1a0bc3385648bd8522c31690c38fda : SUCCESS in 42m 19s\n- zun-tempest-py3-docker-sql-2025-1 https://zuul.opendev.org/t/openstack/build/eab5878ac5b74ef7b04c381870fbb194 : SUCCESS in 43m 43s\n- zun-tempest-py3-docker-sql-2025-2 https://zuul.opendev.org/t/openstack/build/fd788b6805dd4cfd98d65285a14dd9c4 : SUCCESS in 45m 56s\n- zun-tempest-py3-docker-sql-2026-1 https://zuul.opendev.org/t/openstack/build/385a1a9eaeed49d5acef4ce5554f6ad9 : SUCCESS in 43m 01s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/fba27b444e5d48feac6d107438fbc913 : SUCCESS in 22m 37s","accounts_in_message":[],"_revision_number":2},{"id":"9722219f1fe61e21d5eda5c12ad409de97419e75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 14:50:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"13ad3cc3ff103505e2816f318d15ac1caf548ee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 15:22:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05bb6149e22c4891befdb9020ed103a3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/63d79809036444d0b40ab37ed72bcbb2 : SUCCESS in 2m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90d08943825e4a6891c2396bf6dd90a5 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482a96bfab06436d8f91499ec887ff33 : SUCCESS in 2m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/474ea382c7bb46e79af6b58e8c7af910 : SUCCESS in 3m 00s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/8411d52a94dd4dd8b55f5cee0f59ba4f : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":2},{"id":"12492bf67962281a8dbda46075b6b6fe614deb6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 15:22:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3ae2836993c34119f7035bbaed975bca7c0840d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 15:23:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5777cd3dbd3d499e90823d9d04d33d2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc6b4b4d1e804d1bbd9cca2b9e3339e7 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f7849b78145f0d2b18226199b1aefa493b34fd4d","revisions":{"5189fbb4f8230057695bc3eb15a2b10d0f2d13e5":{"kind":"REWORK","_number":1,"created":"2026-08-30 13:59:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/67/1002967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/67/1002967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/67/1002967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/67/1002967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/67/1002967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/67/1002967/1"}}},"commit":{"parents":[{"commit":"390d389576eb031c5a98dea70825b62c6fe15103","subject":"Update requirements for prevent problems with previous releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/390d389576eb031c5a98dea70825b62c6fe15103"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 13:59:49.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 13:59:49.000000000","tz":0},"subject":"Return support for Python 3.10","message":"Return support for Python 3.10\n\nChange-Id: I182acc6582225c6da7278ea10c95ef956f668fd2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5189fbb4f8230057695bc3eb15a2b10d0f2d13e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5189fbb4f8230057695bc3eb15a2b10d0f2d13e5"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"f7849b78145f0d2b18226199b1aefa493b34fd4d":{"kind":"REWORK","_number":2,"created":"2026-08-30 14:00:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/67/1002967/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/67/1002967/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/67/1002967/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/67/1002967/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/67/1002967/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/67/1002967/2"}}},"commit":{"parents":[{"commit":"390d389576eb031c5a98dea70825b62c6fe15103","subject":"Update requirements for prevent problems with previous releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/390d389576eb031c5a98dea70825b62c6fe15103"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 13:59:49.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 14:00:46.000000000","tz":0},"subject":"Return support for Python 3.10","message":"Return support for Python 3.10\n\nChange-Id: I182acc6582225c6da7278ea10c95ef956f668fd2\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f7849b78145f0d2b18226199b1aefa493b34fd4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f7849b78145f0d2b18226199b1aefa493b34fd4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
