)]}'
{"id":"openstack%2Fzun-tempest-plugin~633138","triplet_id":"openstack%2Fzun-tempest-plugin~master~Ic33cbe2b5a3259718562b510344cd1b537fedde1","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ic33cbe2b5a3259718562b510344cd1b537fedde1","subject":"Verify container\u0027s port attributes","status":"MERGED","created":"2019-01-25 04:40:45.000000000","updated":"2019-04-28 05:59:04.000000000","submitted":"2019-04-28 05:59:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"633138-1556431144798-af911147","meta_rev_id":"58caea2e79a77ce6e5d61e2ebfae0e205ed3704f","_number":633138,"virtual_id_number":633138,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-04-28 05:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-04-26 01:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-04-26 01:45:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-28 05:59:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d77c5d1c5d6fe2993c2ec9733ff89ec7a38bf3da","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-01-25 04:40:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78fad19fb2abd49aea5f2eac6f73b05b43f42df0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 05:39:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/1/check/openstack-tox-docs/9646120/html/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/1/check/openstack-tox-pep8/f657955/ : SUCCESS in 4m 56s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/1/check/zun-tempest-docker-sql/b3bc928/ : SUCCESS in 47m 39s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/38/633138/1/check/zun-tempest-py35-docker-sql/696a17d/ : FAILURE in 43m 40s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/1/check/zun-tempest-py35-docker-sql-queens/62e7168/ : FAILURE in 45m 35s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/1/check/zun-tempest-py35-docker-sql-rocky/a004eeb/ : FAILURE in 49m 22s\n- zun-tempest-docker-etcd http://logs.openstack.org/38/633138/1/check/zun-tempest-docker-etcd/65c2ad2/ : SUCCESS in 53m 19s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/38/633138/1/check/zun-tempest-multinode-docker-sql/6ef6f87/ : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":1},{"id":"28cdcd5774d780908d7d48375c300bb8e575efda","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-02-07 03:55:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b669dc960d6396087d6fcff2ea6db17f7b9c66c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-07 04:57:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/1/check/openstack-tox-docs/4a86ecb/html/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/1/check/openstack-tox-pep8/b56d112/ : SUCCESS in 3m 55s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/1/check/zun-tempest-docker-sql/4d7490f/ : SUCCESS in 46m 11s\n- zun-tempest-py35-docker-sql http://logs.openstack.org/38/633138/1/check/zun-tempest-py35-docker-sql/580cbb7/ : SUCCESS in 44m 00s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/1/check/zun-tempest-py35-docker-sql-queens/b54f048/ : FAILURE in 43m 30s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/1/check/zun-tempest-py35-docker-sql-rocky/a25de75/ : FAILURE in 48m 32s\n- zun-tempest-docker-etcd http://logs.openstack.org/38/633138/1/check/zun-tempest-docker-etcd/3f2c19c/ : FAILURE in 47m 06s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/38/633138/1/check/zun-tempest-multinode-docker-sql/73a7ef2/ : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"54e8d7c18fd937655b5f1c60b77dab882ef96047","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-20 16:10:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d50e807fe78091bd97b0f118bf79c52bf3e1fd2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 17:17:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/2/check/openstack-tox-docs/bf3a04f/html/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/2/check/openstack-tox-pep8/435da3e/ : SUCCESS in 3m 44s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/2/check/zun-tempest-docker-sql/fc22cc1/ : SUCCESS in 46m 16s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/2/check/zun-tempest-py35-docker-sql-queens/1a50958/ : FAILURE in 8m 32s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/2/check/zun-tempest-py35-docker-sql-rocky/f1c7054/ : FAILURE in 47m 02s\n- zun-tempest-docker-etcd http://logs.openstack.org/38/633138/2/check/zun-tempest-docker-etcd/9b5b195/ : SUCCESS in 50m 10s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/38/633138/2/check/zun-tempest-multinode-docker-sql/52f19ba/ : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0e164b57f9beb58bb9cb95d75f8b820cf39aef3b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-20 19:14:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d9d75286727c2434cd76e3808e2a99f446a18994","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 20:13:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/3/check/openstack-tox-docs/4c12c2c/html/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/3/check/openstack-tox-pep8/b0f9dcd/ : SUCCESS in 3m 56s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/3/check/zun-tempest-docker-sql/1a28e0e/ : SUCCESS in 50m 47s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/3/check/zun-tempest-py35-docker-sql-queens/d513d3c/ : FAILURE in 49m 03s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/3/check/zun-tempest-py35-docker-sql-rocky/3c5aba6/ : FAILURE in 51m 55s\n- zun-tempest-docker-etcd http://logs.openstack.org/38/633138/3/check/zun-tempest-docker-etcd/25c54b2/ : SUCCESS in 53m 34s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/38/633138/3/check/zun-tempest-multinode-docker-sql/dfb1efb/ : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b11c6bd8368cab748139f7838470ac1f1bcd351e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-20 20:34:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0c037925d102e46e45b1087a529543f3e2b71cf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 22:16:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/4/check/openstack-tox-docs/9cda5bd/html/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/4/check/openstack-tox-pep8/f75f0a8/ : SUCCESS in 3m 53s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/4/check/zun-tempest-docker-sql/d73d900/ : SUCCESS in 48m 19s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/4/check/zun-tempest-py35-docker-sql-queens/4138ba4/ : FAILURE in 46m 51s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/4/check/zun-tempest-py35-docker-sql-rocky/0caf6ef/ : FAILURE in 48m 25s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/38/633138/4/check/zun-tempest-py35-docker-sql-stein/d7fce69/ : SUCCESS in 47m 39s\n- zun-tempest-docker-etcd http://logs.openstack.org/38/633138/4/check/zun-tempest-docker-etcd/a454089/ : SUCCESS in 50m 27s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/38/633138/4/check/zun-tempest-multinode-docker-sql/825e6c3/ : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":4},{"id":"fc9bb804b882cd728ba6a1ca51f0b461a8e04163","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-20 22:23:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"45eb7a05896ea050e612c32183ff2328014376e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 23:37:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/5/check/openstack-tox-docs/e375758/html/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/5/check/openstack-tox-pep8/8d9bd46/ : SUCCESS in 4m 11s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-docker-sql/ef79913/ : SUCCESS in 51m 49s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-queens/5ca53b7/ : SUCCESS in 44m 23s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-rocky/5987b26/ : SUCCESS in 50m 41s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-stein/5b3b34b/ : SUCCESS in 46m 26s\n- zun-tempest-docker-etcd http://logs.openstack.org/38/633138/5/check/zun-tempest-docker-etcd/96addab/ : SUCCESS in 50m 41s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-multinode-docker-sql/f2386b9/ : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":5},{"id":"9d1068599b77c452bfd12d2926339368bd25afd3","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-26 01:45:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"98c7e31ba985ca8eecd42fdd0694cb8e3600bca6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 01:45:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"eef3dd2cd3a8aae6f8d5c5a2b18fda0e17ad0e7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 03:08:48.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/5/gate/openstack-tox-docs/6c62362/html/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/5/gate/openstack-tox-pep8/a7f61cf/ : SUCCESS in 3m 57s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/5/gate/zun-tempest-docker-sql/4dc768e/ : SUCCESS in 46m 57s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/38/633138/5/gate/zun-tempest-py3-docker-sql/ab49b19/ : SUCCESS in 49m 07s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/5/gate/zun-tempest-py35-docker-sql-queens/86e9a4b/ : SUCCESS in 51m 32s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/5/gate/zun-tempest-py35-docker-sql-rocky/5fb9f46/ : TIMED_OUT in 1h 12m 55s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/38/633138/5/gate/zun-tempest-py35-docker-sql-stein/132795f/ : FAILURE in 56m 24s","accounts_in_message":[],"_revision_number":5},{"id":"6af268b98e9d8e18b0871ff50f5b468bddf69cef","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-27 16:04:15.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"00b4c725dacac673d1ccd7e32b0a0f34f9a97f14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 17:07:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/5/check/openstack-tox-docs/07f2de8/html/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/5/check/openstack-tox-pep8/94221fb/ : SUCCESS in 4m 10s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-docker-sql/1fbc160/ : SUCCESS in 50m 52s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-py3-docker-sql/77858e2/ : SUCCESS in 50m 11s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-queens/a7567a5/ : SUCCESS in 46m 39s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-rocky/e286f84/ : SUCCESS in 47m 16s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-stein/44ce63a/ : SUCCESS in 47m 59s\n- zun-tempest-docker-etcd http://logs.openstack.org/38/633138/5/check/zun-tempest-docker-etcd/a65326b/ : SUCCESS in 51m 54s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-multinode-docker-sql/1cd47ac/ : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":5},{"id":"be1f727c970cfa0384313d3dcaf366a594c46faa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 17:08:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9effbc3f27ff3b45afa1993343f3c48905104dcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 18:02:43.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/5/gate/openstack-tox-docs/651b224/html/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/5/gate/openstack-tox-pep8/27e05a5/ : SUCCESS in 6m 53s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/5/gate/zun-tempest-docker-sql/73b79b2/ : SUCCESS in 48m 28s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/38/633138/5/gate/zun-tempest-py3-docker-sql/7f6c805/ : FAILURE in 28m 32s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/5/gate/zun-tempest-py35-docker-sql-queens/bc07181/ : SUCCESS in 44m 05s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/5/gate/zun-tempest-py35-docker-sql-rocky/8d9de02/ : SUCCESS in 46m 30s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/38/633138/5/gate/zun-tempest-py35-docker-sql-stein/a925f9e/ : FAILURE in 28m 51s","accounts_in_message":[],"_revision_number":5},{"id":"5f69fe3742dfd0247d6c02170c29051a40f1c5db","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-27 18:14:10.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3e095cfe946eadea7ea3946dd9b996cea463324e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 19:12:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/5/check/openstack-tox-docs/139d9fb/html/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/5/check/openstack-tox-pep8/8e5209a/ : SUCCESS in 4m 00s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-docker-sql/0dbf0fa/ : SUCCESS in 41m 41s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-py3-docker-sql/8f924dd/ : SUCCESS in 49m 19s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-queens/cefcb95/ : SUCCESS in 48m 38s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-rocky/d6aa6ca/ : SUCCESS in 53m 02s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-stein/2e97cf5/ : SUCCESS in 47m 18s\n- zun-tempest-docker-etcd http://logs.openstack.org/38/633138/5/check/zun-tempest-docker-etcd/889106b/ : SUCCESS in 52m 52s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-multinode-docker-sql/440bde6/ : FAILURE in 43m 19s","accounts_in_message":[],"_revision_number":5},{"id":"378d6edb07f2b409201abc610be10a56351825a6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-28 03:57:49.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f08e7e37c10cac84337ebe8db57fbb44a98d7bc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 05:06:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/5/check/openstack-tox-docs/73bfcba/html/ : SUCCESS in 6m 00s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/5/check/openstack-tox-pep8/4b04725/ : SUCCESS in 5m 37s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-docker-sql/ff2cdf2/ : SUCCESS in 55m 02s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-py3-docker-sql/73788d8/ : SUCCESS in 51m 15s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-queens/8e58240/ : SUCCESS in 47m 55s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-rocky/474c490/ : SUCCESS in 49m 53s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/38/633138/5/check/zun-tempest-py35-docker-sql-stein/a578039/ : SUCCESS in 45m 58s\n- zun-tempest-docker-etcd http://logs.openstack.org/38/633138/5/check/zun-tempest-docker-etcd/bbb9bc1/ : SUCCESS in 51m 20s (non-voting)\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/38/633138/5/check/zun-tempest-multinode-docker-sql/5705c75/ : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":5},{"id":"83a1785cfe80d539ab0e5c79cb04c00b7cbbd3a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 05:06:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"259209ecf31b42f719019f5232a2fdd7a076b2be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 05:59:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/633138/5/gate/openstack-tox-docs/898bc21/html/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/38/633138/5/gate/openstack-tox-pep8/37d57e0/ : SUCCESS in 3m 39s\n- zun-tempest-docker-sql http://logs.openstack.org/38/633138/5/gate/zun-tempest-docker-sql/0f59fad/ : SUCCESS in 49m 05s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/38/633138/5/gate/zun-tempest-py3-docker-sql/0910cfa/ : SUCCESS in 47m 58s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/38/633138/5/gate/zun-tempest-py35-docker-sql-queens/53a25a1/ : SUCCESS in 42m 40s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/38/633138/5/gate/zun-tempest-py35-docker-sql-rocky/027d403/ : SUCCESS in 44m 25s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/38/633138/5/gate/zun-tempest-py35-docker-sql-stein/81b870e/ : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":5},{"id":"58caea2e79a77ce6e5d61e2ebfae0e205ed3704f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 05:59:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3a03fdcc60cfa0a520d9243e70e9052c8af32574","revisions":{"b34e8ff92a6d674b41b5ac511d44bd518e33b433":{"kind":"REWORK","_number":1,"created":"2019-01-25 04:40:45.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/38/633138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/38/633138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/1"}}},"commit":{"parents":[{"commit":"b2313233978525444baa7d8b7f3a7f9aa90f1182","subject":"Test create ipv4 port on dual net","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b2313233978525444baa7d8b7f3a7f9aa90f1182"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-01-25 04:37:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-01-25 04:37:45.000000000","tz":0},"subject":"Verify container\u0027s port attributes","message":"Verify container\u0027s port attributes\n\nVerify several neutron port attributes when the port is created,\nwhen the port is bound to a container, when the container is deleted.\n\nChange-Id: Ic33cbe2b5a3259718562b510344cd1b537fedde1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b34e8ff92a6d674b41b5ac511d44bd518e33b433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b34e8ff92a6d674b41b5ac511d44bd518e33b433"}]},"branch":"refs/heads/master"},"510b28b51cdf94f51ef8157a83ddf1680b69434c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-20 16:10:26.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/38/633138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/38/633138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/2"}}},"commit":{"parents":[{"commit":"1b0fc56dafff369a881bf1c1a57a9d259dd4001f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/1b0fc56dafff369a881bf1c1a57a9d259dd4001f"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-01-25 04:37:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-20 16:10:26.000000000","tz":0},"subject":"Verify container\u0027s port attributes","message":"Verify container\u0027s port attributes\n\nVerify several neutron port attributes when the port is created,\nwhen the port is bound to a container, when the container is deleted.\n\nChange-Id: Ic33cbe2b5a3259718562b510344cd1b537fedde1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/510b28b51cdf94f51ef8157a83ddf1680b69434c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/510b28b51cdf94f51ef8157a83ddf1680b69434c"}]},"branch":"refs/heads/master"},"4093ca75562d1a423c432d92ca5a4ca887ee3d7a":{"kind":"REWORK","_number":3,"created":"2019-04-20 19:14:11.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/38/633138/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/38/633138/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/3"}}},"commit":{"parents":[{"commit":"a2e1b3f26b3286746a974a0394b41ade7d3e1106","subject":"Update tempest documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a2e1b3f26b3286746a974a0394b41ade7d3e1106"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-01-25 04:37:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-20 19:14:05.000000000","tz":0},"subject":"Verify container\u0027s port attributes","message":"Verify container\u0027s port attributes\n\nVerify several neutron port attributes when the port is created,\nwhen the port is bound to a container, when the container is deleted.\n\nChange-Id: Ic33cbe2b5a3259718562b510344cd1b537fedde1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/4093ca75562d1a423c432d92ca5a4ca887ee3d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/4093ca75562d1a423c432d92ca5a4ca887ee3d7a"}]},"branch":"refs/heads/master"},"9d4156aeff0f30d63c938d9757fd2a2a84185874":{"kind":"REWORK","_number":4,"created":"2019-04-20 20:34:43.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/38/633138/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/38/633138/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/4"}}},"commit":{"parents":[{"commit":"a2e1b3f26b3286746a974a0394b41ade7d3e1106","subject":"Update tempest documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a2e1b3f26b3286746a974a0394b41ade7d3e1106"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-01-25 04:37:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-20 20:34:37.000000000","tz":0},"subject":"Verify container\u0027s port attributes","message":"Verify container\u0027s port attributes\n\nVerify several neutron port attributes when the port is created,\nwhen the port is bound to a container, when the container is deleted.\n\nChange-Id: Ic33cbe2b5a3259718562b510344cd1b537fedde1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9d4156aeff0f30d63c938d9757fd2a2a84185874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9d4156aeff0f30d63c938d9757fd2a2a84185874"}]},"branch":"refs/heads/master"},"3a03fdcc60cfa0a520d9243e70e9052c8af32574":{"kind":"REWORK","_number":5,"created":"2019-04-20 22:23:18.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/38/633138/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/38/633138/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/38/633138/5"}}},"commit":{"parents":[{"commit":"a2e1b3f26b3286746a974a0394b41ade7d3e1106","subject":"Update tempest documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a2e1b3f26b3286746a974a0394b41ade7d3e1106"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-01-25 04:37:45.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-20 22:23:08.000000000","tz":0},"subject":"Verify container\u0027s port attributes","message":"Verify container\u0027s port attributes\n\nVerify several neutron port attributes when the port is created,\nwhen the port is bound to a container, when the container is deleted.\n\nChange-Id: Ic33cbe2b5a3259718562b510344cd1b537fedde1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/3a03fdcc60cfa0a520d9243e70e9052c8af32574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/3a03fdcc60cfa0a520d9243e70e9052c8af32574"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
