)]}'
{"id":"openstack%2Fzun-tempest-plugin~656077","triplet_id":"openstack%2Fzun-tempest-plugin~master~Ic709a09c4d42fd310735ecb292e750eb9dc4f71d","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ic709a09c4d42fd310735ecb292e750eb9dc4f71d","subject":"Remove etcd job from gate","status":"MERGED","created":"2019-04-27 17:52:49.000000000","updated":"2019-05-04 18:08:15.000000000","submitted":"2019-05-04 18:08:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656077-1556993295197-5260f087","meta_rev_id":"b94c183978620c195a7abded8f201b9cffb9223e","_number":656077,"virtual_id_number":656077,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-05-04 18:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-05-04 17:10:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-05-04 17:10:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 18:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e0c64d5d7492ca9d5820f86f36096ed0d6bee36","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-27 17:52:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"288c55127ddde579a21182939b2f89c67835bdbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 19:15:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/77/656077/1/check/openstack-tox-docs/2f8fe24/html/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/77/656077/1/check/openstack-tox-pep8/01efde1/ : SUCCESS in 5m 16s\n- zun-tempest-docker-sql http://logs.openstack.org/77/656077/1/check/zun-tempest-docker-sql/bc17dfe/ : SUCCESS in 53m 07s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/77/656077/1/check/zun-tempest-py3-docker-sql/f874f17/ : FAILURE in 51m 34s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/77/656077/1/check/zun-tempest-py35-docker-sql-queens/f9778ea/ : SUCCESS in 45m 22s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/77/656077/1/check/zun-tempest-py35-docker-sql-rocky/05836e2/ : SUCCESS in 50m 00s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/77/656077/1/check/zun-tempest-py35-docker-sql-stein/fe043e5/ : SUCCESS in 48m 16s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/77/656077/1/check/zun-tempest-multinode-docker-sql/abb077f/ : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a596d4bf0366347176b6b09bbaf9523bfde12b31","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-04-28 03:58:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"05d60b711adc08b07ff496bc88574d3f884cf4c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 05:03:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/77/656077/1/check/openstack-tox-docs/dd2f255/html/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/77/656077/1/check/openstack-tox-pep8/70b8c87/ : SUCCESS in 3m 39s\n- zun-tempest-docker-sql http://logs.openstack.org/77/656077/1/check/zun-tempest-docker-sql/3968cfa/ : SUCCESS in 48m 59s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/77/656077/1/check/zun-tempest-py3-docker-sql/a399fd3/ : SUCCESS in 46m 38s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/77/656077/1/check/zun-tempest-py35-docker-sql-queens/6742844/ : SUCCESS in 45m 57s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/77/656077/1/check/zun-tempest-py35-docker-sql-rocky/b54cfd1/ : SUCCESS in 47m 49s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/77/656077/1/check/zun-tempest-py35-docker-sql-stein/d388f9d/ : SUCCESS in 45m 43s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/77/656077/1/check/zun-tempest-multinode-docker-sql/6e95dea/ : SUCCESS in 59m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e289933f79ce84e20ab76e572c4cc17e5cd10c1e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-04 17:10:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc3c4315bd0bb5639d4871933dec215ef1a5db9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 17:11:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a4c3cc5b41bc77af017d9ec2caaea5bf33e8ca1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 18:08:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/77/656077/1/gate/openstack-tox-docs/3323766/html/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/77/656077/1/gate/openstack-tox-pep8/a150b73/ : SUCCESS in 3m 50s\n- zun-tempest-docker-sql http://logs.openstack.org/77/656077/1/gate/zun-tempest-docker-sql/260c18b/ : SUCCESS in 51m 38s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/77/656077/1/gate/zun-tempest-py3-docker-sql/99913e0/ : SUCCESS in 47m 57s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/77/656077/1/gate/zun-tempest-py35-docker-sql-queens/04ed998/ : SUCCESS in 46m 51s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/77/656077/1/gate/zun-tempest-py35-docker-sql-rocky/f774491/ : SUCCESS in 48m 58s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/77/656077/1/gate/zun-tempest-py35-docker-sql-stein/842f441/ : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b94c183978620c195a7abded8f201b9cffb9223e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 18:08:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d309e43e61842cd3ca1b1b48aaa53161a66b5dc0","revisions":{"d309e43e61842cd3ca1b1b48aaa53161a66b5dc0":{"kind":"REWORK","_number":1,"created":"2019-04-27 17:52:49.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/77/656077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/77/656077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/77/656077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/77/656077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/77/656077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/77/656077/1"}}},"commit":{"parents":[{"commit":"e4c84a942859c05a7d84dcc5a6d3120c14e068ef","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/e4c84a942859c05a7d84dcc5a6d3120c14e068ef"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-27 17:52:00.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-04-27 17:52:49.000000000","tz":0},"subject":"Remove etcd job from gate","message":"Remove etcd job from gate\n\nThe etcd backend is duplicated for removal. This commit removes\nthe etcd job.\n\nChange-Id: Ic709a09c4d42fd310735ecb292e750eb9dc4f71d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d309e43e61842cd3ca1b1b48aaa53161a66b5dc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d309e43e61842cd3ca1b1b48aaa53161a66b5dc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
