)]}'
{"id":"openstack%2Fzun-tempest-plugin~657304","triplet_id":"openstack%2Fzun-tempest-plugin~master~I299ae26ba6de5ed438579dd73d59416a54b0be98","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I299ae26ba6de5ed438579dd73d59416a54b0be98","subject":"Remove assertion of resource constraint","status":"MERGED","created":"2019-05-06 02:56:15.000000000","updated":"2019-05-08 03:42:42.000000000","submitted":"2019-05-08 03:42:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657304-1557286962967-b00d5b3e","meta_rev_id":"6bcf4fa1f859c034128aa91df9207b90ccca3da6","_number":657304,"virtual_id_number":657304,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-05-08 03:42:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-05-07 21:36:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-05-07 21:36:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-08 03:42:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0a91e8f5358bdae49aad5798b79585b39d7a08d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-06 02:56:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7f3ab03ca90cd27defbb80c7cafd6725c24bb70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 04:08:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/657304/1/check/openstack-tox-docs/a106d8e/html/ : SUCCESS in 4m 24s\n- openstack-tox-pep8 http://logs.openstack.org/04/657304/1/check/openstack-tox-pep8/577fdae/ : SUCCESS in 3m 56s\n- zun-tempest-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-docker-sql/ddc44b5/ : SUCCESS in 44m 02s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-py3-docker-sql/d0e7d2b/ : SUCCESS in 41m 57s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-queens/f2205f3/ : FAILURE in 43m 25s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-rocky/a537296/ : FAILURE in 45m 35s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-stein/e3e862b/ : FAILURE in 44m 34s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-multinode-docker-sql/5ab8beb/ : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":1},{"id":"67461ce55428c6655510344abe8f063db62d0434","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-06 05:10:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d14fe87ea243da8cc4fa34d9624e23fc3aab4f26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 06:33:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/657304/1/check/openstack-tox-docs/09eed2e/html/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/04/657304/1/check/openstack-tox-pep8/5c77ba3/ : SUCCESS in 5m 23s\n- zun-tempest-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-docker-sql/7487f31/ : SUCCESS in 59m 16s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-py3-docker-sql/a3fe3d3/ : SUCCESS in 46m 32s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-queens/1d4f7b5/ : SUCCESS in 44m 56s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-rocky/f8c64bb/ : FAILURE in 56m 42s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-stein/cd09bac/ : FAILURE in 39m 46s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-multinode-docker-sql/076d0fe/ : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b500c8dc1618611ff620000a1a90ef43eb749c5f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-07 11:46:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8df39a7758d8114bdfe7b9c6a080b0561fc29234","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 12:52:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/04/657304/1/check/openstack-tox-docs/126d177/html/ : SUCCESS in 6m 26s\n- openstack-tox-pep8 http://logs.openstack.org/04/657304/1/check/openstack-tox-pep8/43d4ecd/ : SUCCESS in 4m 09s\n- zun-tempest-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-docker-sql/160b5fb/ : SUCCESS in 51m 13s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-py3-docker-sql/1be0582/ : SUCCESS in 48m 56s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-queens/ee0ccf0/ : SUCCESS in 49m 02s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-rocky/2cc121f/ : SUCCESS in 56m 18s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-stein/e667603/ : SUCCESS in 47m 03s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-multinode-docker-sql/49faa2c/ : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a27205d077a1388fb9f3c5f4783732a3a04df8a3","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-07 21:36:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09db95bfdfad2e721bca9e30b450cdc04ae8a676","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 21:37:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3621329904e1f6d454e85ba75f037aa860a5bea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 22:49:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/657304/1/gate/openstack-tox-docs/c2bde5d/html/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/04/657304/1/gate/openstack-tox-pep8/3d2409c/ : SUCCESS in 4m 04s\n- zun-tempest-docker-sql http://logs.openstack.org/04/657304/1/gate/zun-tempest-docker-sql/f4fb997/ : SUCCESS in 54m 16s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/04/657304/1/gate/zun-tempest-py3-docker-sql/c181457/ : FAILURE in 13m 28s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/04/657304/1/gate/zun-tempest-py35-docker-sql-queens/c8e9b72/ : SUCCESS in 44m 55s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/04/657304/1/gate/zun-tempest-py35-docker-sql-rocky/7a1bc02/ : FAILURE in 17m 12s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/04/657304/1/gate/zun-tempest-py35-docker-sql-stein/21bf268/ : FAILURE in 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1fcb6323342b2029ade29e16cefe2f00f103c681","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-05-08 01:41:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"03b86cdd3cdf62a965f755a02d7fd5bffebb588b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 02:47:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/04/657304/1/check/openstack-tox-docs/3cd456a/html/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/04/657304/1/check/openstack-tox-pep8/d1ca60e/ : SUCCESS in 3m 56s\n- zun-tempest-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-docker-sql/e102564/ : SUCCESS in 50m 50s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-py3-docker-sql/7ce45a1/ : SUCCESS in 50m 37s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-queens/28e57d9/ : SUCCESS in 47m 49s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-rocky/ea1b348/ : SUCCESS in 49m 23s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/04/657304/1/check/zun-tempest-py35-docker-sql-stein/beeda3a/ : SUCCESS in 47m 37s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/04/657304/1/check/zun-tempest-multinode-docker-sql/35bb5b5/ : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4b3f7d0c25358f819b2d13076dd05ac7c83bd3f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 02:47:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4811a5137e6412b607d58a375e0a5672f70d48ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 03:42:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/04/657304/1/gate/openstack-tox-docs/c78d64a/html/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/04/657304/1/gate/openstack-tox-pep8/295d01a/ : SUCCESS in 4m 07s\n- zun-tempest-docker-sql http://logs.openstack.org/04/657304/1/gate/zun-tempest-docker-sql/53dd713/ : SUCCESS in 45m 26s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/04/657304/1/gate/zun-tempest-py3-docker-sql/54d44d7/ : SUCCESS in 54m 17s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/04/657304/1/gate/zun-tempest-py35-docker-sql-queens/793f36d/ : SUCCESS in 43m 58s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/04/657304/1/gate/zun-tempest-py35-docker-sql-rocky/a7fa575/ : SUCCESS in 46m 20s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/04/657304/1/gate/zun-tempest-py35-docker-sql-stein/1ec402a/ : SUCCESS in 46m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6bcf4fa1f859c034128aa91df9207b90ccca3da6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 03:42:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6327701485d832c9419de80737233dfb6750c267","revisions":{"6327701485d832c9419de80737233dfb6750c267":{"kind":"REWORK","_number":1,"created":"2019-05-06 02:56:15.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/04/657304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/04/657304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/04/657304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/04/657304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/04/657304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/04/657304/1"}}},"commit":{"parents":[{"commit":"4639700c1f20903b7c9be830a5219320703a91e6","subject":"Merge \"Remove etcd job from gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/4639700c1f20903b7c9be830a5219320703a91e6"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-06 02:52:15.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-05-06 02:55:47.000000000","tz":0},"subject":"Remove assertion of resource constraint","message":"Remove assertion of resource constraint\n\nThis assertion doesn\u0027t work well in multinode gate. This commit\nremoves them temporarily. Later, we will revisit this logic.\nPerhaps, we can re-introduce it in a non-tempest pipeline.\n\nChange-Id: I299ae26ba6de5ed438579dd73d59416a54b0be98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6327701485d832c9419de80737233dfb6750c267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6327701485d832c9419de80737233dfb6750c267"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
