)]}'
{"id":"openstack%2Fzun-tempest-plugin~662898","triplet_id":"openstack%2Fzun-tempest-plugin~master~Ief6c48fdbd54ace79a37dca6a096d716a771b79d","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"Ief6c48fdbd54ace79a37dca6a096d716a771b79d","subject":"Sync Sphinx requirement","status":"MERGED","created":"2019-06-04 01:44:08.000000000","updated":"2019-06-05 03:04:37.000000000","submitted":"2019-06-05 03:04:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662898-1559703877306-717f2137","meta_rev_id":"fea346db646a019b27e3c0f926aa7c1deaa6ce39","_number":662898,"virtual_id_number":662898,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-06-05 03:04:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-06-05 02:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2019-06-05 01:37:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-06-05 02:06:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 01:37:18.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-06-05 02:06:38.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2019-06-05 03:04:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"806eb7fc921b3082570e9ae25ae241cff726fa85","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-04 01:44:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73aae88486a5bf366cd7ed7375672c84626db54b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 02:50:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/98/662898/1/check/requirements-check/e72f531/ : SUCCESS in 3m 04s\n- openstack-tox-docs http://logs.openstack.org/98/662898/1/check/openstack-tox-docs/9fffd76/html/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/98/662898/1/check/openstack-tox-pep8/ccfb02b/ : SUCCESS in 4m 19s\n- zun-tempest-docker-sql http://logs.openstack.org/98/662898/1/check/zun-tempest-docker-sql/490a2de/ : SUCCESS in 46m 52s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/98/662898/1/check/zun-tempest-py3-docker-sql/c12826e/ : SUCCESS in 46m 23s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/98/662898/1/check/zun-tempest-py35-docker-sql-queens/6823fce/ : SUCCESS in 50m 17s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/98/662898/1/check/zun-tempest-py35-docker-sql-rocky/0583c41/ : SUCCESS in 49m 37s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/98/662898/1/check/zun-tempest-py35-docker-sql-stein/32f300f/ : SUCCESS in 46m 33s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/98/662898/1/check/zun-tempest-multinode-docker-sql/eb14f05/ : SUCCESS in 1h 00m 02s","accounts_in_message":[],"_revision_number":1},{"id":"aced5a02b298e666e3173ad506f6f27d42d6ca9e","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-05 01:37:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"17850a4568bd5735e84559313c38ae66b5736365","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-05 02:06:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e86e953e6bdf14cd2f6c4b49d6b07e245cdfd842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:06:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c5313e2c921d5e1ea66dbf84aba9306d6b6432f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 03:04:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fea346db646a019b27e3c0f926aa7c1deaa6ce39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 03:04:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/98/662898/1/gate/requirements-check/cbd8189/ : SUCCESS in 3m 29s\n- openstack-tox-docs http://logs.openstack.org/98/662898/1/gate/openstack-tox-docs/597fe2f/html/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/98/662898/1/gate/openstack-tox-pep8/6ed2936/ : SUCCESS in 4m 51s\n- zun-tempest-docker-sql http://logs.openstack.org/98/662898/1/gate/zun-tempest-docker-sql/f3a4066/ : SUCCESS in 52m 29s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/98/662898/1/gate/zun-tempest-py3-docker-sql/983cf87/ : SUCCESS in 50m 12s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/98/662898/1/gate/zun-tempest-py35-docker-sql-queens/257a777/ : SUCCESS in 47m 04s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/98/662898/1/gate/zun-tempest-py35-docker-sql-rocky/672a5ef/ : SUCCESS in 41m 30s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/98/662898/1/gate/zun-tempest-py35-docker-sql-stein/f1a8ec7/ : SUCCESS in 46m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5dd082207e41ac8118291ad27fecc79a0482b9e","revisions":{"f5dd082207e41ac8118291ad27fecc79a0482b9e":{"kind":"REWORK","_number":1,"created":"2019-06-04 01:44:08.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/98/662898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/98/662898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/98/662898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/98/662898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/98/662898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/98/662898/1"}}},"commit":{"parents":[{"commit":"841c7d3793559a7741e2b83f7ce1b56f8d119977","subject":"Merge \"Add test for exposing container ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/841c7d3793559a7741e2b83f7ce1b56f8d119977"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-06-04 01:42:55.000000000","tz":480},"committer":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-06-04 01:44:04.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\nSync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n\nChange-Id: Ief6c48fdbd54ace79a37dca6a096d716a771b79d\nCloses-Bug: #1831517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f5dd082207e41ac8118291ad27fecc79a0482b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f5dd082207e41ac8118291ad27fecc79a0482b9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
