)]}'
{"id":"openstack%2Fzun-tempest-plugin~664145","triplet_id":"openstack%2Fzun-tempest-plugin~master~I5eb04993f39a4098cde4293fccd07f8458a1cd5d","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I5eb04993f39a4098cde4293fccd07f8458a1cd5d","subject":"Remove tests for docker container","status":"MERGED","created":"2019-06-08 21:56:59.000000000","updated":"2019-06-30 00:09:26.000000000","submitted":"2019-06-30 00:09:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664145-1561853366431-e6a449f5","meta_rev_id":"ee7f513c8bd724aecf9a0cd65d585ceb150df234","_number":664145,"virtual_id_number":664145,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-06-30 00:09:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-06-29 23:15:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-06-29 23:15:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-30 00:09:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60c506829af4f1ffaa37e5a364a81f9acdd7c64a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-08 21:56:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de329d79a8845f36e2f40fd17f4563efff76bd1b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-08 22:06:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"108ef39d276867288852befd525fb98dc337a12b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 23:11:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/45/664145/2/check/openstack-tox-docs/3de4e8b/html/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/45/664145/2/check/openstack-tox-pep8/9bd13f8/ : SUCCESS in 5m 26s\n- zun-tempest-docker-sql http://logs.openstack.org/45/664145/2/check/zun-tempest-docker-sql/e02491f/ : FAILURE in 46m 53s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/45/664145/2/check/zun-tempest-py3-docker-sql/9f25fc6/ : FAILURE in 43m 12s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/45/664145/2/check/zun-tempest-py35-docker-sql-queens/50c0a43/ : SUCCESS in 49m 25s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/45/664145/2/check/zun-tempest-py35-docker-sql-rocky/6b5d455/ : FAILURE in 49m 27s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/45/664145/2/check/zun-tempest-py35-docker-sql-stein/e9e0243/ : FAILURE in 46m 09s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/45/664145/2/check/zun-tempest-multinode-docker-sql/c507877/ : FAILURE in 1h 00m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e55810a250f6b8d3b3410548e60dce21f0472588","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-09 16:37:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f65d2d466c6cec133509ccc59e62e4ddc9bf5fe4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 17:52:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/45/664145/3/check/openstack-tox-docs/6b5d39f/html/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/45/664145/3/check/openstack-tox-pep8/fbd0a38/ : SUCCESS in 4m 12s\n- zun-tempest-docker-sql http://logs.openstack.org/45/664145/3/check/zun-tempest-docker-sql/8065314/ : SUCCESS in 52m 47s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/45/664145/3/check/zun-tempest-py3-docker-sql/c873ac3/ : SUCCESS in 46m 54s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/45/664145/3/check/zun-tempest-py35-docker-sql-queens/2c2a567/ : SUCCESS in 52m 00s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/45/664145/3/check/zun-tempest-py35-docker-sql-rocky/b49d6f7/ : SUCCESS in 50m 06s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/45/664145/3/check/zun-tempest-py35-docker-sql-stein/3825ec4/ : SUCCESS in 49m 43s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/45/664145/3/check/zun-tempest-multinode-docker-sql/3379642/ : SUCCESS in 1h 09m 29s","accounts_in_message":[],"_revision_number":3},{"id":"abbb3d02d3dcec3baa4669f0a7ecbf3c55ead418","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-29 23:15:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a75fcb063fe42d07ca8b984548adca38a78449ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 23:15:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c3c8db880996b8cd42484360f2c4503377439fb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 00:09:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ee7f513c8bd724aecf9a0cd65d585ceb150df234","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 00:09:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/45/664145/3/gate/openstack-tox-docs/f30e485/html/ : SUCCESS in 2m 31s\n- openstack-tox-pep8 http://logs.openstack.org/45/664145/3/gate/openstack-tox-pep8/decbd79/ : SUCCESS in 2m 44s\n- zun-tempest-docker-sql http://logs.openstack.org/45/664145/3/gate/zun-tempest-docker-sql/059f75a/ : SUCCESS in 48m 10s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/45/664145/3/gate/zun-tempest-py3-docker-sql/feeeeb3/ : SUCCESS in 42m 39s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/45/664145/3/gate/zun-tempest-py35-docker-sql-queens/224ab04/ : SUCCESS in 44m 45s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/45/664145/3/gate/zun-tempest-py35-docker-sql-rocky/124cc88/ : SUCCESS in 48m 58s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/45/664145/3/gate/zun-tempest-py35-docker-sql-stein/c7dca3b/ : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"552948142d4f4fb1f0f47e54a7e9fca31d353a6c","revisions":{"dffe9875396e50b2565eb7d3e5725bc2f7e477fe":{"kind":"REWORK","_number":1,"created":"2019-06-08 21:56:59.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/45/664145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/45/664145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/1"}}},"commit":{"parents":[{"commit":"f5dd082207e41ac8118291ad27fecc79a0482b9e","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f5dd082207e41ac8118291ad27fecc79a0482b9e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-08 21:52:24.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-08 21:56:59.000000000","tz":0},"subject":"[WIP] Remove tests for docker container","message":"[WIP] Remove tests for docker container\n\nTests for verifying docker containers is removed from tempest\ntest suite and re-introduce in fullstack tests.\nThis is because tempest tests are supposed to be a pure API tests.\nTouching docker containers are beyond the scope of tempest tests\nso we split them out.\n\nDepends-On: https://review.opendev.org/#/c/658702/\nChange-Id: I5eb04993f39a4098cde4293fccd07f8458a1cd5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/dffe9875396e50b2565eb7d3e5725bc2f7e477fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/dffe9875396e50b2565eb7d3e5725bc2f7e477fe"}]},"branch":"refs/heads/master"},"dacd792e947401f95e2bdd58475c9f7fe9dcaed0":{"kind":"REWORK","_number":2,"created":"2019-06-08 22:06:37.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/45/664145/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/45/664145/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/2"}}},"commit":{"parents":[{"commit":"f5dd082207e41ac8118291ad27fecc79a0482b9e","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f5dd082207e41ac8118291ad27fecc79a0482b9e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-08 21:52:24.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-08 22:06:32.000000000","tz":0},"subject":"[WIP] Remove tests for docker container","message":"[WIP] Remove tests for docker container\n\nTests for verifying docker containers is removed from tempest\ntest suite and re-introduce in fullstack tests.\nThis is because tempest tests are supposed to be a pure API tests.\nTouching docker containers are beyond the scope of tempest tests\nso we split them out.\n\nDepends-On: https://review.opendev.org/#/c/658702/\nChange-Id: I5eb04993f39a4098cde4293fccd07f8458a1cd5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/dacd792e947401f95e2bdd58475c9f7fe9dcaed0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/dacd792e947401f95e2bdd58475c9f7fe9dcaed0"}]},"branch":"refs/heads/master"},"552948142d4f4fb1f0f47e54a7e9fca31d353a6c":{"kind":"REWORK","_number":3,"created":"2019-06-09 16:37:27.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/45/664145/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/45/664145/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/45/664145/3"}}},"commit":{"parents":[{"commit":"f5dd082207e41ac8118291ad27fecc79a0482b9e","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f5dd082207e41ac8118291ad27fecc79a0482b9e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-08 21:52:24.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-09 16:37:15.000000000","tz":0},"subject":"Remove tests for docker container","message":"Remove tests for docker container\n\nTests for verifying docker containers is removed from tempest\ntest suite and re-introduce in fullstack tests.\nThis is because tempest tests are supposed to be a pure API tests.\nTouching docker containers are beyond the scope of tempest tests\nso we split them out.\n\nDepends-On: https://review.opendev.org/#/c/658702/\nChange-Id: I5eb04993f39a4098cde4293fccd07f8458a1cd5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/552948142d4f4fb1f0f47e54a7e9fca31d353a6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/552948142d4f4fb1f0f47e54a7e9fca31d353a6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
