)]}'
{"id":"openstack%2Fzun-tempest-plugin~667489","triplet_id":"openstack%2Fzun-tempest-plugin~master~I071f18177d6a073334e7d4adbc0d0657e7ae6f01","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I071f18177d6a073334e7d4adbc0d0657e7ae6f01","subject":"[DNM] Test CI","status":"ABANDONED","created":"2019-06-26 02:01:21.000000000","updated":"2019-06-27 02:40:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7285b78f7d11a8e04178508bd599ebba20fcca72","_number":667489,"virtual_id_number":667489,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-26 03:09:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 03:09:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4134785bbc789d91e03a961990c24c8e86ede820","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-26 02:01:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a36934afa297bad59994d2c749d0b652adb7eee8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 03:09:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/89/667489/1/check/test-release-openstack/87611e1/ : SUCCESS in 1m 44s\n- openstack-tox-docs http://logs.openstack.org/89/667489/1/check/openstack-tox-docs/53cf46f/html/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/89/667489/1/check/openstack-tox-pep8/eb3a823/ : SUCCESS in 2m 48s\n- zun-tempest-docker-sql http://logs.openstack.org/89/667489/1/check/zun-tempest-docker-sql/b69c723/ : FAILURE in 42m 25s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/89/667489/1/check/zun-tempest-py3-docker-sql/5c794b1/ : FAILURE in 40m 51s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/89/667489/1/check/zun-tempest-py35-docker-sql-queens/a179426/ : SUCCESS in 49m 23s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/89/667489/1/check/zun-tempest-py35-docker-sql-rocky/2dde545/ : FAILURE in 45m 24s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/89/667489/1/check/zun-tempest-py35-docker-sql-stein/72576c4/ : FAILURE in 45m 51s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/89/667489/1/check/zun-tempest-multinode-docker-sql/2a728e7/ : FAILURE in 1h 01m 21s","accounts_in_message":[],"_revision_number":1},{"id":"7285b78f7d11a8e04178508bd599ebba20fcca72","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-27 02:40:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b59c5c8283288ad81c0a5d2659bd2dd749399aa3","revisions":{"b59c5c8283288ad81c0a5d2659bd2dd749399aa3":{"kind":"REWORK","_number":1,"created":"2019-06-26 02:01:21.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/89/667489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/89/667489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/89/667489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/89/667489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/89/667489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/89/667489/1"}}},"commit":{"parents":[{"commit":"f5dd082207e41ac8118291ad27fecc79a0482b9e","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f5dd082207e41ac8118291ad27fecc79a0482b9e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-26 02:01:08.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-26 02:01:08.000000000","tz":0},"subject":"[DNM] Test CI","message":"[DNM] Test CI\n\nChange-Id: I071f18177d6a073334e7d4adbc0d0657e7ae6f01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b59c5c8283288ad81c0a5d2659bd2dd749399aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/b59c5c8283288ad81c0a5d2659bd2dd749399aa3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
