)]}'
{"id":"openstack%2Fzun-tempest-plugin~667491","triplet_id":"openstack%2Fzun-tempest-plugin~master~I7528bc4b562e0701c40c7e8b482d5a50783abcdf","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I7528bc4b562e0701c40c7e8b482d5a50783abcdf","subject":"Disable SSL certificate validation","status":"MERGED","created":"2019-06-26 02:04:44.000000000","updated":"2019-06-27 02:25:14.000000000","submitted":"2019-06-27 02:25:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667491-1561602314126-e8d46088","meta_rev_id":"10d91df3a991feb8dea3058d3faac114423eef91","_number":667491,"virtual_id_number":667491,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-06-27 02:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-06-27 01:26:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-06-27 01:26:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 02:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"647d610bf7cd0d0b46928e5dc5db4c28b8f0af85","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-26 02:04:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b494de2c53e67085a20ba1e7afdce63e5b22bc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 03:11:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/667491/1/check/openstack-tox-docs/16ff45b/html/ : SUCCESS in 2m 42s\n- openstack-tox-pep8 http://logs.openstack.org/91/667491/1/check/openstack-tox-pep8/5d48e38/ : SUCCESS in 2m 42s\n- zun-tempest-docker-sql http://logs.openstack.org/91/667491/1/check/zun-tempest-docker-sql/cf9b665/ : FAILURE in 45m 28s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/91/667491/1/check/zun-tempest-py3-docker-sql/c9ee026/ : FAILURE in 45m 23s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/91/667491/1/check/zun-tempest-py35-docker-sql-queens/663aea5/ : SUCCESS in 46m 13s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/91/667491/1/check/zun-tempest-py35-docker-sql-rocky/6dff48a/ : FAILURE in 46m 11s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/91/667491/1/check/zun-tempest-py35-docker-sql-stein/72a370a/ : FAILURE in 44m 34s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/91/667491/1/check/zun-tempest-multinode-docker-sql/3112813/ : FAILURE in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"cb5902d82dd95da86732e1ea6a488440b6d45000","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-26 04:12:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a34a72f0ad301e8f1dde880737156e9963677b10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 05:30:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/667491/2/check/openstack-tox-docs/ad2a6aa/html/ : SUCCESS in 2m 41s\n- openstack-tox-pep8 http://logs.openstack.org/91/667491/2/check/openstack-tox-pep8/1b42237/ : SUCCESS in 4m 00s\n- zun-tempest-docker-sql http://logs.openstack.org/91/667491/2/check/zun-tempest-docker-sql/b6a5392/ : SUCCESS in 48m 39s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/91/667491/2/check/zun-tempest-py3-docker-sql/b15ff8e/ : SUCCESS in 49m 18s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/91/667491/2/check/zun-tempest-py35-docker-sql-queens/a751dd4/ : SUCCESS in 50m 13s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/91/667491/2/check/zun-tempest-py35-docker-sql-rocky/acc06b6/ : SUCCESS in 50m 22s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/91/667491/2/check/zun-tempest-py35-docker-sql-stein/5bbf632/ : FAILURE in 48m 53s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/91/667491/2/check/zun-tempest-multinode-docker-sql/24fb135/ : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a2ce4a74875068c43b6d8c8fb4149b5b639b5b1a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-26 13:13:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"536a93fac3aad13662160316c7122355758f99b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 14:29:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/667491/2/check/openstack-tox-docs/ca9c78a/html/ : SUCCESS in 2m 46s\n- openstack-tox-pep8 http://logs.openstack.org/91/667491/2/check/openstack-tox-pep8/08e128c/ : SUCCESS in 3m 00s\n- zun-tempest-docker-sql http://logs.openstack.org/91/667491/2/check/zun-tempest-docker-sql/aa7aff3/ : SUCCESS in 48m 34s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/91/667491/2/check/zun-tempest-py3-docker-sql/38b5492/ : SUCCESS in 46m 02s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/91/667491/2/check/zun-tempest-py35-docker-sql-queens/45f3651/ : SUCCESS in 47m 54s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/91/667491/2/check/zun-tempest-py35-docker-sql-rocky/12dfe10/ : SUCCESS in 55m 38s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/91/667491/2/check/zun-tempest-py35-docker-sql-stein/6ba2167/ : SUCCESS in 52m 33s\n- zun-tempest-multinode-docker-sql http://logs.openstack.org/91/667491/2/check/zun-tempest-multinode-docker-sql/1edc3b4/ : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":2},{"id":"732258d53ec63f3708569e1a62fe5e3d4a005346","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-06-27 01:26:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9dbeb4faead933811c29aa14992794cf0eb1099c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 01:26:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b2254ecc4545c053c6a1fa3765fd91f9a049e932","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 02:25:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/667491/2/gate/openstack-tox-docs/27850fd/html/ : SUCCESS in 2m 53s\n- openstack-tox-pep8 http://logs.openstack.org/91/667491/2/gate/openstack-tox-pep8/002cfb1/ : SUCCESS in 2m 36s\n- zun-tempest-docker-sql http://logs.openstack.org/91/667491/2/gate/zun-tempest-docker-sql/32e801b/ : SUCCESS in 53m 10s\n- zun-tempest-py3-docker-sql http://logs.openstack.org/91/667491/2/gate/zun-tempest-py3-docker-sql/8dda47f/ : SUCCESS in 50m 07s\n- zun-tempest-py35-docker-sql-queens http://logs.openstack.org/91/667491/2/gate/zun-tempest-py35-docker-sql-queens/2a6580f/ : SUCCESS in 48m 53s\n- zun-tempest-py35-docker-sql-rocky http://logs.openstack.org/91/667491/2/gate/zun-tempest-py35-docker-sql-rocky/197f5fc/ : SUCCESS in 51m 04s\n- zun-tempest-py35-docker-sql-stein http://logs.openstack.org/91/667491/2/gate/zun-tempest-py35-docker-sql-stein/627dbcb/ : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":2},{"id":"10d91df3a991feb8dea3058d3faac114423eef91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 02:25:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d4387a34c7b67f190db227556b0ddf480011c6cf","revisions":{"bf9b4c47a366ba6a492fee01439df57d68aa5039":{"kind":"REWORK","_number":1,"created":"2019-06-26 02:04:44.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/91/667491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/91/667491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/91/667491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/91/667491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/91/667491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/91/667491/1"}}},"commit":{"parents":[{"commit":"f5dd082207e41ac8118291ad27fecc79a0482b9e","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f5dd082207e41ac8118291ad27fecc79a0482b9e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-26 02:02:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-26 02:02:22.000000000","tz":0},"subject":"[WIP] Disable SSL certificate validation","message":"[WIP] Disable SSL certificate validation\n\nStarting from urllib3 1.25.x, it validate SSL certificate by default,\nwhich failed our CI.\n\nChange-Id: I7528bc4b562e0701c40c7e8b482d5a50783abcdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/bf9b4c47a366ba6a492fee01439df57d68aa5039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/bf9b4c47a366ba6a492fee01439df57d68aa5039"}]},"branch":"refs/heads/master"},"d4387a34c7b67f190db227556b0ddf480011c6cf":{"kind":"REWORK","_number":2,"created":"2019-06-26 04:12:37.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/91/667491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/91/667491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/91/667491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/91/667491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/91/667491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/91/667491/2"}}},"commit":{"parents":[{"commit":"f5dd082207e41ac8118291ad27fecc79a0482b9e","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f5dd082207e41ac8118291ad27fecc79a0482b9e"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-26 02:02:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-06-26 04:12:27.000000000","tz":0},"subject":"Disable SSL certificate validation","message":"Disable SSL certificate validation\n\nStarting from urllib3 1.25.x, it validate SSL certificate by default,\nwhich failed our CI.\n\nChange-Id: I7528bc4b562e0701c40c7e8b482d5a50783abcdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d4387a34c7b67f190db227556b0ddf480011c6cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/d4387a34c7b67f190db227556b0ddf480011c6cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
