)]}'
{"id":"openstack%2Fzun-tempest-plugin~674210","triplet_id":"openstack%2Fzun-tempest-plugin~master~I43a278afd810498606c4ce5bc2ca50079380599e","project":"openstack/zun-tempest-plugin","branch":"master","topic":"openstackdocstheme","hashtags":[],"change_id":"I43a278afd810498606c4ce5bc2ca50079380599e","subject":"Bump the openstackdocstheme extension to 1.20","status":"MERGED","created":"2019-08-02 07:08:48.000000000","updated":"2019-08-18 16:59:31.000000000","submitted":"2019-08-18 16:59:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674210-1566147571548-ff059232","meta_rev_id":"e2af104557259b7feebdfdcbe8aa09640b201090","_number":674210,"virtual_id_number":674210,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-08-18 16:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-12 01:35:17.000000000","_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-08-05 13:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-08-05 13:27:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 13:27:56.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2019-08-18 16:59:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77c0e35125fa11ab38f437a91f8d80658c0b1692","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-02 07:08:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29d3010dd4dbfd60220cbea602c2bb1da8cba35f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 09:09:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/10/674210/1/check/requirements-check/4eebf2c/ : SUCCESS in 3m 01s\n- openstack-tox-docs https://logs.opendev.org/10/674210/1/check/openstack-tox-docs/686ccdf/html/ : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://logs.opendev.org/10/674210/1/check/openstack-tox-pep8/dcce218/ : SUCCESS in 4m 14s\n- zun-tempest-docker-sql https://logs.opendev.org/10/674210/1/check/zun-tempest-docker-sql/36101e7/ : SUCCESS in 50m 30s\n- zun-tempest-py3-docker-sql https://logs.opendev.org/10/674210/1/check/zun-tempest-py3-docker-sql/a99a423/ : SUCCESS in 55m 57s\n- zun-tempest-py35-docker-sql-queens https://logs.opendev.org/10/674210/1/check/zun-tempest-py35-docker-sql-queens/9bc03d2/ : SUCCESS in 55m 14s\n- zun-tempest-py35-docker-sql-rocky https://logs.opendev.org/10/674210/1/check/zun-tempest-py35-docker-sql-rocky/dfdebf1/ : SUCCESS in 48m 54s\n- zun-tempest-py35-docker-sql-stein https://logs.opendev.org/10/674210/1/check/zun-tempest-py35-docker-sql-stein/c202e5b/ : SUCCESS in 49m 12s\n- zun-tempest-multinode-docker-sql https://logs.opendev.org/10/674210/1/check/zun-tempest-multinode-docker-sql/6f6fb06/ : SUCCESS in 1h 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9850f3a3d3743257f5b3e965e65adf3ab026256a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-03 15:55:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2251af123bd3cb0ecec52c28d5a6236544ff07f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 16:49:36.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"b456dae0b6bb535897f32aab2e0de02ae4f857ed","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-05 01:38:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"db90c38f44a2d68041e3817076074e03d34e25d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 02:42:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/10/674210/2/check/requirements-check/d60c79d/ : SUCCESS in 1m 48s\n- openstack-tox-docs https://logs.opendev.org/10/674210/2/check/openstack-tox-docs/03b2b33/html/ : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://logs.opendev.org/10/674210/2/check/openstack-tox-pep8/a86dfc7/ : SUCCESS in 2m 31s\n- zun-tempest-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-docker-sql/4ee8198/ : SUCCESS in 48m 32s\n- zun-tempest-py3-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-py3-docker-sql/6f27b02/ : SUCCESS in 42m 35s\n- zun-tempest-py35-docker-sql-queens https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-queens/f28a256/ : SUCCESS in 46m 38s\n- zun-tempest-py35-docker-sql-rocky https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-rocky/919da84/ : SUCCESS in 48m 44s\n- zun-tempest-py35-docker-sql-stein https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-stein/b454e1b/ : SUCCESS in 41m 30s\n- zun-tempest-multinode-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-multinode-docker-sql/39fd025/ : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":2},{"id":"5a31467d9c316fd41d9b7bccb7cd4ae2b09260ca","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-05 13:27:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d4b447b7be47fbf94d6e15130c2928ba4583172c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:28:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"412b1e9d3e5bec442152cce714a57d287089b869","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:33:38.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/10/674210/2/gate/requirements-check/58975a5/ : SUCCESS in 3m 14s\n- openstack-tox-docs https://logs.opendev.org/10/674210/2/gate/openstack-tox-docs/b271f51/html/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://logs.opendev.org/10/674210/2/gate/openstack-tox-pep8/c9dddf5/ : SUCCESS in 4m 23s\n- zun-tempest-docker-sql https://logs.opendev.org/10/674210/2/gate/zun-tempest-docker-sql/330fdf9/ : SUCCESS in 50m 20s\n- zun-tempest-py3-docker-sql https://logs.opendev.org/10/674210/2/gate/zun-tempest-py3-docker-sql/7857067/ : SUCCESS in 46m 15s\n- zun-tempest-py35-docker-sql-queens https://logs.opendev.org/10/674210/2/gate/zun-tempest-py35-docker-sql-queens/420dbb2/ : SUCCESS in 50m 35s\n- zun-tempest-py35-docker-sql-rocky https://logs.opendev.org/10/674210/2/gate/zun-tempest-py35-docker-sql-rocky/c5b9625/ : SUCCESS in 59m 45s\n- zun-tempest-py35-docker-sql-stein https://logs.opendev.org/10/674210/2/gate/zun-tempest-py35-docker-sql-stein/776594b/ : FAILURE in 49m 04s","accounts_in_message":[],"_revision_number":2},{"id":"861089455c9f9e1ec889614c90e2e5ebc2f563a9","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-06 01:44:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad69970d4c0e1de5a1222274ae1afd2f9dbcd81b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 02:43:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/10/674210/2/check/requirements-check/0ca2a0a/ : SUCCESS in 2m 04s\n- openstack-tox-docs https://logs.opendev.org/10/674210/2/check/openstack-tox-docs/0987703/html/ : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://logs.opendev.org/10/674210/2/check/openstack-tox-pep8/6da39f0/ : SUCCESS in 2m 45s\n- zun-tempest-docker-sql finger://ze03.openstack.org/41af0a11a5fa46acb9ef73bb5e5c58d0 : POST_FAILURE in 45m 03s\n- zun-tempest-py3-docker-sql finger://ze02.openstack.org/13822d2302ea40faa40da20ade650683 : POST_FAILURE in 42m 13s\n- zun-tempest-py35-docker-sql-queens finger://ze02.openstack.org/de0f3b9b1673467799dff7e099f2e242 : POST_FAILURE in 41m 29s\n- zun-tempest-py35-docker-sql-rocky finger://ze02.openstack.org/484d1d0d1aca418cb08c1a86a500a2be : POST_FAILURE in 46m 22s\n- zun-tempest-py35-docker-sql-stein finger://ze01.openstack.org/cc59d97d59e94c1b8a66570fb2ac78bc : POST_FAILURE in 42m 54s\n- zun-tempest-multinode-docker-sql finger://ze03.openstack.org/aba85fcd7fd3409ebe722856783db2cc : POST_FAILURE in 56m 45s","accounts_in_message":[],"_revision_number":2},{"id":"dc8b410cb202fcb605260da59d53cef91aee3ed1","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-09 02:09:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"96dd2023b96dd29d1f3b432dc1d6f74c86ec0ae0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 03:22:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/10/674210/2/check/requirements-check/2b1dc02/ : SUCCESS in 2m 52s\n- openstack-tox-docs https://logs.opendev.org/10/674210/2/check/openstack-tox-docs/1760639/html/ : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://logs.opendev.org/10/674210/2/check/openstack-tox-pep8/e827a08/ : SUCCESS in 4m 43s\n- zun-tempest-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-docker-sql/afcf09a/ : SUCCESS in 49m 30s\n- zun-tempest-py3-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-py3-docker-sql/1d24b9b/ : SUCCESS in 1h 03m 48s\n- zun-tempest-py35-docker-sql-queens https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-queens/e7cf771/ : FAILURE in 55m 34s\n- zun-tempest-py35-docker-sql-rocky https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-rocky/e16e65f/ : FAILURE in 1h 05m 35s\n- zun-tempest-py35-docker-sql-stein https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-stein/dd8058b/ : FAILURE in 1h 10m 17s\n- zun-tempest-multinode-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-multinode-docker-sql/478a7ba/ : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f55d778f6271e381a8808fb162eeb0af3c73f463","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-09 08:01:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d4126dff9bf51068ccc53e06c51e9502e4929ae0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:32:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/10/674210/2/check/requirements-check/6c48e8b/ : SUCCESS in 1m 58s\n- openstack-tox-docs https://logs.opendev.org/10/674210/2/check/openstack-tox-docs/77503a1/html/ : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://logs.opendev.org/10/674210/2/check/openstack-tox-pep8/4b6201c/ : SUCCESS in 3m 06s\n- zun-tempest-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-docker-sql/9ab7aad/ : SUCCESS in 1h 01m 22s\n- zun-tempest-py3-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-py3-docker-sql/0a67f2a/ : SUCCESS in 55m 05s\n- zun-tempest-py35-docker-sql-queens https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-queens/7ad3f49/ : SUCCESS in 49m 57s\n- zun-tempest-py35-docker-sql-rocky https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-rocky/c9e5a99/ : SUCCESS in 50m 38s\n- zun-tempest-py35-docker-sql-stein https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-stein/f96f9d4/ : SUCCESS in 48m 17s\n- zun-tempest-multinode-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-multinode-docker-sql/3b59215/ : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":2},{"id":"9602e5ac81337a06598b3dab38d519264e7da5c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:32:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6dd93319c3e4b1b9060c4c3c1debcf6bd4212faf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 10:48:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/10/674210/2/gate/requirements-check/203019f/ : SUCCESS in 3m 22s\n- openstack-tox-docs https://logs.opendev.org/10/674210/2/gate/openstack-tox-docs/507a45a/html/ : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://logs.opendev.org/10/674210/2/gate/openstack-tox-pep8/9c8cd13/ : SUCCESS in 3m 15s\n- zun-tempest-docker-sql https://logs.opendev.org/10/674210/2/gate/zun-tempest-docker-sql/a457c81/ : SUCCESS in 59m 26s\n- zun-tempest-py3-docker-sql https://logs.opendev.org/10/674210/2/gate/zun-tempest-py3-docker-sql/1708944/ : SUCCESS in 48m 56s\n- zun-tempest-py35-docker-sql-queens https://logs.opendev.org/10/674210/2/gate/zun-tempest-py35-docker-sql-queens/f403fc7/ : SUCCESS in 49m 13s\n- zun-tempest-py35-docker-sql-rocky https://logs.opendev.org/10/674210/2/gate/zun-tempest-py35-docker-sql-rocky/a20e873/ : SUCCESS in 54m 30s\n- zun-tempest-py35-docker-sql-stein https://logs.opendev.org/10/674210/2/gate/zun-tempest-py35-docker-sql-stein/09bc681/ : FAILURE in 59m 00s","accounts_in_message":[],"_revision_number":2},{"id":"36b108ff76c550bfac51d3fe159fe2a4180d5d79","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-12 01:35:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"97120eda1e9721e9e8a02efbae454473052efbcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 02:33:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/10/674210/2/check/requirements-check/93c8399/ : SUCCESS in 1m 50s\n- openstack-tox-docs https://logs.opendev.org/10/674210/2/check/openstack-tox-docs/d65d760/html/ : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://logs.opendev.org/10/674210/2/check/openstack-tox-pep8/de9025c/ : SUCCESS in 2m 42s\n- zun-tempest-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-docker-sql/4627c5e/ : SUCCESS in 55m 37s\n- zun-tempest-py3-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-py3-docker-sql/1635f8b/ : SUCCESS in 50m 09s\n- zun-tempest-py35-docker-sql-queens https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-queens/f61aaff/ : SUCCESS in 42m 50s\n- zun-tempest-py35-docker-sql-rocky https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-rocky/0ffb9da/ : SUCCESS in 47m 20s\n- zun-tempest-py35-docker-sql-stein https://logs.opendev.org/10/674210/2/check/zun-tempest-py35-docker-sql-stein/57b5092/ : FAILURE in 53m 35s\n- zun-tempest-multinode-docker-sql https://logs.opendev.org/10/674210/2/check/zun-tempest-multinode-docker-sql/1ed5dad/ : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9cf3358ae720fc2b1c6a1bc83e1232d1c9ebd350","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-18 14:51:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2447f540443ac7d0b9823bbd41dd48d4d28f00cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 16:10:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2b98555a55b4c84a4bc1bd425ed7ace : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f34bebc3928f41a0b7992193fbf76fef : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b16ad9126b114cc5bd7a03e8a03f6bc3 : SUCCESS in 3m 45s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/9cbd1c743836480391cd94b96220d8ea : SUCCESS in 49m 00s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/9eec67cdb1b943edaa30b6ca65c320c4 : SUCCESS in 45m 08s\n- zun-tempest-py35-docker-sql-queens https://zuul.opendev.org/t/openstack/build/e0435f4c328e4d77a7af2ddefb456937 : SUCCESS in 44m 40s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/753e66c53a494708b4306f8446660977 : SUCCESS in 47m 33s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/09447eecdc7241dea970b43126b30420 : SUCCESS in 49m 29s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/bc67841360664f57941cf84226dcdc5a : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e25f0f4a0a7a2c82c10e447d003bace071040f1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 16:10:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8e07c860b0b2c4e2997abbf9fcaa53ca05856daa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 16:59:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cbb1dc56d2fd6931cfe759a73a24baeb15f73b89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 16:59:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c682568458d4fe58368750ced61ab8a : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c05e6b5cb1654a7b8a69cb12fc57f5c5 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bea524276c940649a321b542f925844 : SUCCESS in 2m 35s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/6dc4e68046854388a1438f4ebc3d15ec : SUCCESS in 41m 30s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ba3f4cd134884a4eb9d06e74f61ebe2a : SUCCESS in 43m 12s\n- zun-tempest-py35-docker-sql-queens https://zuul.opendev.org/t/openstack/build/9c1e6ac25db8467e8fc1e92b346c6a09 : SUCCESS in 43m 58s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/6151b00bc8d74f97b4b9b520948be03a : SUCCESS in 44m 51s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/0e22270ea6c0429583f5336069b148a4 : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de8893e91fd04ca0f3fe6062b4acc6d54d6ed45b","revisions":{"94939439b29abd85edbfea3e105bcd61bf30055d":{"kind":"REWORK","_number":1,"created":"2019-08-02 07:08:48.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/10/674210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/10/674210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/10/674210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/10/674210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/10/674210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/10/674210/1"}}},"commit":{"parents":[{"commit":"fbc14c99bbffd183be52599e508f7df902d2f9b5","subject":"Merge \"Remove tests for docker container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/fbc14c99bbffd183be52599e508f7df902d2f9b5"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 07:08:48.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 07:08:51.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: I43a278afd810498606c4ce5bc2ca50079380599e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/94939439b29abd85edbfea3e105bcd61bf30055d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/94939439b29abd85edbfea3e105bcd61bf30055d"}]},"branch":"refs/heads/master"},"de8893e91fd04ca0f3fe6062b4acc6d54d6ed45b":{"kind":"REWORK","_number":2,"created":"2019-08-05 01:38:33.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/10/674210/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/10/674210/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/10/674210/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/10/674210/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/10/674210/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/10/674210/2"}}},"commit":{"parents":[{"commit":"f2564f08d3dcaa8e3b5529849002950b090d592d","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f2564f08d3dcaa8e3b5529849002950b090d592d"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 07:08:48.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-05 01:38:38.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: I43a278afd810498606c4ce5bc2ca50079380599e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/de8893e91fd04ca0f3fe6062b4acc6d54d6ed45b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/de8893e91fd04ca0f3fe6062b4acc6d54d6ed45b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
