)]}'
{"id":"openstack%2Fzun-tempest-plugin~677094","triplet_id":"openstack%2Fzun-tempest-plugin~master~Iee1f53c905863eb842535090a2bb67c3babf9b05","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"Iee1f53c905863eb842535090a2bb67c3babf9b05","subject":"Skip tests if glance is not available","status":"MERGED","created":"2019-08-18 15:46:47.000000000","updated":"2019-08-18 19:50:16.000000000","submitted":"2019-08-18 19:50:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677094-1566157816562-9a06ee28","meta_rev_id":"2889045ec544e1ca70eaae442fb5aa2814f4f797","_number":677094,"virtual_id_number":677094,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2019-08-18 19:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2019-08-18 19:05:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2019-08-18 19:05:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-18 19:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"115b33aab8b95c8c43e4b2b471651c5682cf7c93","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-18 15:46:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb83bcf1b3fd0ad225779c88634f46b883183428","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 16:52:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea2aef105e8343d78e3d28e5d256d5ad : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a20aac1212e408a94063bc74236c5a0 : SUCCESS in 2m 58s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/6c7c3581ff494abda3b0cce6355b7129 : SUCCESS in 51m 51s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/f74908dca02d4e3981f9ac385ae542ba : SUCCESS in 48m 02s\n- zun-tempest-py35-docker-sql-queens https://zuul.opendev.org/t/openstack/build/779f53e5a4474c2a909b58e156a2382c : SUCCESS in 44m 43s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/84c5ca280c91439691b290ac1c807d8a : SUCCESS in 51m 07s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/fc1be6d8ee0143059a30672cb8e350c5 : SUCCESS in 49m 23s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ae23e5fabe5c44e58891017cc80c0792 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c1bd39f612796ccdc2f323ac7b02a9a82d735ac9","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2019-08-18 19:05:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac9c152e94594f5bef6ce436868a5072ec3f3f4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 19:05:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"629fba82f682be6b30d610998fdcdbcacd1a2136","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 19:50:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2889045ec544e1ca70eaae442fb5aa2814f4f797","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 19:50:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd094a21f594bbfb341a45d392bf391 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63560f2dc632490087e4a5aeb6f5b2a2 : SUCCESS in 3m 30s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/30442ace7c764a9d9cccd82abf6d0769 : SUCCESS in 42m 15s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/eff758cc3f944af791d1968508fbcd32 : SUCCESS in 39m 17s\n- zun-tempest-py35-docker-sql-queens https://zuul.opendev.org/t/openstack/build/d8232939966a4437855cbcb144f53278 : SUCCESS in 37m 18s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/f07600d905d34bb8be4928a1add530d5 : SUCCESS in 42m 01s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/ee82e3bc39324fcca2ef7984a2ca94ea : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71037184185613f9c11544050398c8ec3ceb531d","revisions":{"71037184185613f9c11544050398c8ec3ceb531d":{"kind":"REWORK","_number":1,"created":"2019-08-18 15:46:47.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/94/677094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/94/677094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/94/677094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/94/677094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/94/677094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/94/677094/1"}}},"commit":{"parents":[{"commit":"f2564f08d3dcaa8e3b5529849002950b090d592d","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/f2564f08d3dcaa8e3b5529849002950b090d592d"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-08-18 15:43:32.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2019-08-18 15:46:46.000000000","tz":0},"subject":"Skip tests if glance is not available","message":"Skip tests if glance is not available\n\nThere are some tests requiring glance service. If glance service\nis not available, we want to skip those tests instead of failing\nthem. The following are the affected tests:\n* test_commit_container\n* test_run_container_with_image_driver_glance\n\nChange-Id: Iee1f53c905863eb842535090a2bb67c3babf9b05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/71037184185613f9c11544050398c8ec3ceb531d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/71037184185613f9c11544050398c8ec3ceb531d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
