)]}'
{"id":"openstack%2Fzun-tempest-plugin~692481","triplet_id":"openstack%2Fzun-tempest-plugin~master~I729ae1a6c106cbc126d6d5849ed2f86673a29b4e","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I729ae1a6c106cbc126d6d5849ed2f86673a29b4e","subject":"Update \"python setup.py build_sphinx\" to build_sphinx","status":"MERGED","created":"2019-11-01 03:39:18.000000000","updated":"2020-05-22 16:55:34.000000000","submitted":"2020-05-22 16:53:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692481-1590166405350-8638809e","meta_rev_id":"8a6e45479aee1b78a40377f7a8267e2c93dae976","_number":692481,"virtual_id_number":692481,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"date":"2020-05-22 14:32:43.000000000","_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 16:53:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-05-22 02:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-05-22 02:44:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 02:44:47.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-05-22 16:53:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4841a01620e618d4933acebdd1d30a72c2cb93e4","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-11-01 03:39:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcce447e4203fa1d878e64ec9739be2382d9cd95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 05:23:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8073ea295a8843acaf279889b0a44d4a : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d73388db904e11a445b5ce6d9a77fc : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adb7f921d60a4f2c938d2537f58b557c : SUCCESS in 2m 59s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/d5dac64e9a564e189e4af496f4e48de2 : SUCCESS in 45m 38s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/474ef6a5a7d44ae49757f96a8b622ac2 : SUCCESS in 41m 25s\n- zun-tempest-py35-docker-sql-queens https://zuul.opendev.org/t/openstack/build/48e159a978be4be39d0d3820656a20c4 : SUCCESS in 40m 07s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/dce12eec1bcb420da9bd3419c600c6c9 : SUCCESS in 47m 44s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/5cdeee60fb89491c88838e3002be3f0a : SUCCESS in 41m 40s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/5e99b0ee246b44e6af746352567f1f08 : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5d579bfd1f96e2e712afe4c8f4f9a6189662cbae","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-11-01 05:36:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e736773f12aa67e7a15d0d047f08e2f8054f039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 06:39:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0a0e59ae0d84dffa7b780e68a05b7cc : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3bf099bc70f4fe1833d9f0cbc7ef9e3 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/394b15734f744ded93481b6d4a7c293e : SUCCESS in 2m 51s\n- zun-tempest-docker-sql https://zuul.opendev.org/t/openstack/build/1767cf4569264d13931ced236055c86f : SUCCESS in 50m 24s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/2dad17cfe4f0460aaf77f431d20dfdaf : SUCCESS in 46m 21s\n- zun-tempest-py35-docker-sql-queens https://zuul.opendev.org/t/openstack/build/a8c3cc2595f54ba4b6bd4b1fb27941c5 : SUCCESS in 41m 44s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/fb05d2491ba846f9ad14c3c684a7df74 : SUCCESS in 44m 48s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/d7a0c9dac1d0421e9a3ce21177548ae3 : SUCCESS in 45m 32s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/551976299cf24727bfc949ec42287427 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":2},{"id":"aa5d64b77390a28d20aabb8d530604e831871d23","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-21 01:28:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fadc941b91c99398b12fda5bc6ad7b16b8872085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 02:27:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/637f79c75ee740cca6eac6339020a69c : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e01b29cc8fda4afa8881715ed3e87740 : SUCCESS in 3m 37s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/0701623aeef744fc9b2df5d42242d122 : SUCCESS in 52m 22s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/e611208a0f424d289399506541aa0d99 : SUCCESS in 45m 04s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/a89bbe87d1a9466680f2371dd6b17df5 : SUCCESS in 42m 46s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/d12a787d00a24fcbb679fba8778def40 : SUCCESS in 41m 18s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/07fece4cb2e3471687bd02f01ee53aac : SUCCESS in 57m 27s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2acfb3282c3d44f483cdfe0be8337415 : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":3},{"id":"da8eef9a88db07edbafb260a0cbe2cf7d07f1634","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-22 02:44:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fa7ace57e20056130693f78875f1e6461c62fca1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 02:44:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6213ee164805f0c572bffbf31886bb7013639ef7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 03:43:11.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67188b9bc1c9494dbc20241d3b2c92da : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb7c66d6d0504d2eba62709ed0cf1144 : SUCCESS in 3m 33s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/5b7d4c0bb60f45de923d78ddefa2b14c : SUCCESS in 51m 28s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/947a97a5cd21464295dd7fcb2d37667c : SUCCESS in 52m 47s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/83bcbb42a8004d8f939638ae54ec3b23 : SUCCESS in 48m 42s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/0b105736d1614a368c16e03e0e4fbd3d : FAILURE in 40m 47s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/d63e35a0eff84c08ac577495c6d368e5 : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":3},{"id":"1b5e671b0e6a5b8c3226840beda7356fd6b58741","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-05-22 14:32:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"525f31126b18aa7825c6e0ef576899e01f231ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 15:47:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8149503aa865414097dac80c3fcf30a1 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbe40e59d9ac401282d22128c32aa94e : SUCCESS in 5m 44s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/8db304cbbca54784ab095ef6cea5ba98 : SUCCESS in 54m 25s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/c32c016773a246d79e26f37491aedee8 : SUCCESS in 54m 42s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/3959cce29ca647d289d8f600516cccc3 : SUCCESS in 55m 22s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/f6f0f9d9033e4f69ba6d32c8e8ca5d1a : SUCCESS in 50m 48s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/637fa54cdcce455dbfa5987010980ce9 : SUCCESS in 52m 18s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/2ed4a60f9c81469ea164db2968d9f666 : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":3},{"id":"677c3ef76cd833258323da18bf0fa80b4a936020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 15:47:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6e7e1be037d2198fc74d55e585bf7cb32289fa1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 16:53:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3fa0fa4caacf8144f03476625c1455a696aa7dc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 16:53:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7922ef37584d4255b605356560939ff9 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51016105f21a436aacc3edc1b054bd9c : SUCCESS in 6m 54s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/5adf10061c4142819912dc8057f9f35b : SUCCESS in 57m 01s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/22a15cbe21d34e3e93103e151bf638ed : SUCCESS in 53m 28s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/1d56c8595ab3468aa72fa37b69291f09 : SUCCESS in 49m 15s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/ad286ba606804a6ba837f63f1ead2fe1 : SUCCESS in 54m 34s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/86f80d7d464548b39aba54608fd9ed09 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":3},{"id":"8a6e45479aee1b78a40377f7a8267e2c93dae976","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 16:55:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d1af4113ba45f29d3452f26aceb632 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"01d03bb71edd5661b233feddb9afd52be76e5949","revisions":{"2c83b266a4cfdc417e31a1919db73a3481f2d767":{"kind":"REWORK","_number":1,"created":"2019-11-01 03:39:18.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/81/692481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/81/692481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/1"}}},"commit":{"parents":[{"commit":"0d639a465ea289781d5a3cb96fb283a85a284007","subject":"Merge \"Add tempest test for init containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/0d639a465ea289781d5a3cb96fb283a85a284007"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-11-01 03:39:07.000000000","tz":480},"committer":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-11-01 03:39:07.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nChange-Id: I729ae1a6c106cbc126d6d5849ed2f86673a29b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/2c83b266a4cfdc417e31a1919db73a3481f2d767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/2c83b266a4cfdc417e31a1919db73a3481f2d767"}]},"branch":"refs/heads/master"},"682ba87b159e562bb629b09c91fb3f7149ad870e":{"kind":"REWORK","_number":2,"created":"2019-11-01 05:36:44.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/81/692481/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/81/692481/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/2"}}},"commit":{"parents":[{"commit":"0d639a465ea289781d5a3cb96fb283a85a284007","subject":"Merge \"Add tempest test for init containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/0d639a465ea289781d5a3cb96fb283a85a284007"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-11-01 03:39:07.000000000","tz":480},"committer":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-11-01 05:35:33.000000000","tz":480},"subject":"Update \"python setup.py build_sphinx\" to build_sphinx","message":"Update \"python setup.py build_sphinx\" to build_sphinx\n\n1. Update \"python setup.py build_sphinx\" to build_sphinx\n2. Keeping going with docs\n\nChange-Id: I729ae1a6c106cbc126d6d5849ed2f86673a29b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/682ba87b159e562bb629b09c91fb3f7149ad870e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/682ba87b159e562bb629b09c91fb3f7149ad870e"}]},"branch":"refs/heads/master"},"01d03bb71edd5661b233feddb9afd52be76e5949":{"kind":"REWORK","_number":3,"created":"2020-05-21 01:28:45.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/81/692481/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/81/692481/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/81/692481/3"}}},"commit":{"parents":[{"commit":"5df462dcfa971b743b1305cbf5d44bec585ee436","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5df462dcfa971b743b1305cbf5d44bec585ee436"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-11-01 03:39:07.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-05-21 01:28:27.000000000","tz":480},"subject":"Update \"python setup.py build_sphinx\" to build_sphinx","message":"Update \"python setup.py build_sphinx\" to build_sphinx\n\n1. Update \"python setup.py build_sphinx\" to build_sphinx\n2. Keeping going with docs\n\nChange-Id: I729ae1a6c106cbc126d6d5849ed2f86673a29b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/01d03bb71edd5661b233feddb9afd52be76e5949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/01d03bb71edd5661b233feddb9afd52be76e5949"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
