)]}'
{"id":"openstack%2Fzun-tempest-plugin~706717","triplet_id":"openstack%2Fzun-tempest-plugin~master~Iceb2ebd82b87e258b760e0eb3977665ef30f5dac","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"Iceb2ebd82b87e258b760e0eb3977665ef30f5dac","subject":"Remove stable/queens job","status":"MERGED","created":"2020-02-09 18:47:47.000000000","updated":"2020-02-12 05:49:12.000000000","submitted":"2020-02-12 05:47:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706717-1581486440274-4f5f8f67","meta_rev_id":"a4035cc8e2a3fea20bbd04ff5c07012343bef77d","_number":706717,"virtual_id_number":706717,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-12 05:47:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-02-12 04:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-02-12 04:44:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 05:47:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a1eba1fc4033a065081853795096d7b949586e3","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-09 18:47:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94a8351326643d6227bef7118415407c2b9d82a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 19:54:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eae0cf4790d43f299104b3b54ac8e1f : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b02277e89c348ffbfe0a20df6170126 : SUCCESS in 4m 10s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/aeedcc4235f949549137e97705a29d18 : SUCCESS in 52m 10s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/6b6b4ade84cb4e04aef502392110676a : SUCCESS in 47m 07s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/5385ec8c80f344eda42275963831a637 : SUCCESS in 1h 01m 42s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ae40ea58e10c4e988cae92dbccf9ead6 : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d35a690b93e823d03260b8a3f4c2ff9ed4453135","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-12 04:44:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa8f294710822428fde98190e14fc13beca96a44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 04:44:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42ad7ad9437f10904c1e1691d3f9266786a1000c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 05:47:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83420ef7ff1348c8b74bd9a8c59f32a1 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c399e7cfaf45559b212d320fe4eba6 : SUCCESS in 5m 16s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/7672e1c6ac854e6698f554efa10d07bd : SUCCESS in 49m 35s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/98b997aaf79748259081a7880edc3dd3 : SUCCESS in 43m 10s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/99614b3a5a4b4c7bb3a22effc0f3a5d0 : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f3eddfc90c111382a7fb2502bdebce557335b7e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 05:47:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4035cc8e2a3fea20bbd04ff5c07012343bef77d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 05:49:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c71fa0b21d42e3bb2b182d8f74d94e : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b836e0048e07df585f4ee5526ed5bbe78edffa7","revisions":{"9b836e0048e07df585f4ee5526ed5bbe78edffa7":{"kind":"REWORK","_number":1,"created":"2020-02-09 18:47:47.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/17/706717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/17/706717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/17/706717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/17/706717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/17/706717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/17/706717/1"}}},"commit":{"parents":[{"commit":"0d639a465ea289781d5a3cb96fb283a85a284007","subject":"Merge \"Add tempest test for init containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/0d639a465ea289781d5a3cb96fb283a85a284007"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-09 18:46:04.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-09 18:46:04.000000000","tz":0},"subject":"Remove stable/queens job","message":"Remove stable/queens job\n\nThe stable/queens branch enters extended maintenance phrase\n\nChange-Id: Iceb2ebd82b87e258b760e0eb3977665ef30f5dac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9b836e0048e07df585f4ee5526ed5bbe78edffa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9b836e0048e07df585f4ee5526ed5bbe78edffa7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
