)]}'
{"id":"openstack%2Fzun-tempest-plugin~721073","triplet_id":"openstack%2Fzun-tempest-plugin~master~I3d2e0459ba4aedf21c01d7725fe5ab266f9f4da3","project":"openstack/zun-tempest-plugin","branch":"master","topic":"docs-cleanup","hashtags":[],"change_id":"I3d2e0459ba4aedf21c01d7725fe5ab266f9f4da3","subject":"Update docs building","status":"MERGED","created":"2020-04-19 08:25:00.000000000","updated":"2020-04-19 16:43:19.000000000","submitted":"2020-04-19 16:41:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721073-1587314515756-eff0fd40","meta_rev_id":"7d27def576355eaeeb4f8a1e8b893847db69dc3f","_number":721073,"virtual_id_number":721073,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-19 16:41:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-04-19 15:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-04-19 12:36:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-04-19 15:55:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-19 12:36:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-19 15:55:32.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-04-19 16:41:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b571682f54b028047099a7a19e34ece82ad6081","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-19 08:25:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"049b38c97fac368c217f2344396efe4beacd243d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 09:29:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e6e0c217a376411b9e8e46f61654be09 : SUCCESS in 2m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/20b7be66ec1c4d3da70dee3b34708005 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169a24e692104485aee03a701e75ec51 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18661d716aca4f15b3dd3a98ab0f213d : SUCCESS in 3m 28s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/74de67b1dfeb47c18a256060512a55e7 : SUCCESS in 46m 53s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/1cc78e8d0c9e4628bc4643c984a65c68 : SUCCESS in 50m 00s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/c7d5bc473dc44285bf692f91bfe1ae87 : SUCCESS in 46m 29s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/7e70707a39ea4823be95e3d211925610 : SUCCESS in 46m 44s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/8bc8aa56ecf548f9976d2b10a66e27db : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4d61861183a2179b50636c39f6a293169cda43c9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-19 12:36:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"38659405e133cec4dbe8015edb136d44502a80de","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-04-19 15:55:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1235f30334c0c4a586f209065df71e076a3fc751","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 15:55:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc9c16a658a921fadfaa63baad604f71b4a8b9dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 16:41:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ba1da9f77ba4b9db64a541a975b973d : SUCCESS in 2m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/77d53fe5fab84a039ed21bd43c9dd102 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd19835424fc4957ac3aa200d5eacf96 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0882a6da086c4ef4b216e26978ae610d : SUCCESS in 3m 33s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/c5660913d848428187c92488fb5b2a30 : SUCCESS in 44m 21s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/3109c9c985ee42059b7c334275b8e353 : SUCCESS in 41m 55s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/d137013a94de433d9aad02f83e275b58 : SUCCESS in 37m 05s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/0fef808883ba47f6b1a79d5b34a38fe5 : SUCCESS in 45m 21s","accounts_in_message":[],"_revision_number":1},{"id":"623990a07219d88b0516809cc907bf9a273233f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 16:41:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"13f4aa326eb3ed657571f85d220b3c3a363cc21f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 16:43:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8b662eb73a4eabb8a50f82704d7115 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9043347888f26d5c9dc804f758597de2efea0181","revisions":{"9043347888f26d5c9dc804f758597de2efea0181":{"kind":"REWORK","_number":1,"created":"2020-04-19 08:25:00.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/73/721073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/73/721073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/73/721073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/73/721073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/73/721073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/73/721073/1"}}},"commit":{"parents":[{"commit":"a5ff9442ec6eb40b5edd1f37d8f192f563fa77db","subject":"Add test case for requested host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/a5ff9442ec6eb40b5edd1f37d8f192f563fa77db"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 08:23:38.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 08:24:33.000000000","tz":120},"subject":"Update docs building","message":"Update docs building\n\nUpdate docs building:\n* Switch to sphinx-build\n* Update requirements for Sphinx and openstackdocstheme for\n  python 3\n* Remove unneeded doc and translation sections from setup.cfg\n* Remove install_command, it\u0027s unneeded, the default is fine.\n\nChange-Id: I3d2e0459ba4aedf21c01d7725fe5ab266f9f4da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9043347888f26d5c9dc804f758597de2efea0181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9043347888f26d5c9dc804f758597de2efea0181"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
