)]}'
{"id":"openstack%2Fzun-tempest-plugin~725192","triplet_id":"openstack%2Fzun-tempest-plugin~master~I1c9dd70d2d3319ad9012620fe8478316f6a940d2","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I1c9dd70d2d3319ad9012620fe8478316f6a940d2","subject":"Remove six","status":"MERGED","created":"2020-05-04 06:44:55.000000000","updated":"2020-05-05 14:37:37.000000000","submitted":"2020-05-05 14:35:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725192-1588689352526-b9db8533","meta_rev_id":"34b7723a3cf8be3f182a11bad4ef2a8d8c6764e8","_number":725192,"virtual_id_number":725192,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-05 14:35:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-05-05 13:33:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-05-05 13:33:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 13:33:28.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-05-05 14:35:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4f969cc8c52e07f41d2856a9cf26da432a5b7b8","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-04 06:44:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd9f2ec17c233fed93c5aa99c5e073819ba692c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 07:40:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7bca4038c5bb4fb2a7c802c8fe2f275e : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6d7a99160f8490bb026ce0438fe4520 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af7aa6a9ccf460f9673665a68a4ce91 : FAILURE in 3m 44s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/b4f6b1b24bee4e71b686bc94b77aaf03 : FAILURE in 29m 34s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/06eedefd6b934147ac5118b2086bccdb : FAILURE in 45m 30s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/24847aa22bc1484ba006eb7cbc863df1 : FAILURE in 43m 46s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/ff43fe6d2c8441a48e257044f0c5fbec : FAILURE in 48m 52s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/17bdb2dd60d04febae26e13cb0225020 : FAILURE in 38m 47s","accounts_in_message":[],"_revision_number":1},{"id":"26543505a2cef7ce03c1beb8f81432e6cba8e3d7","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-04 07:50:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19a4cec7758a91536c523c5a8681ea9181c21660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 08:44:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/25740f72e4144627ab09f850529e0867 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a1d03282a347a8b7e6c46b8004cf8c : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5400f15958ca4b338866e8fc6e95c12f : SUCCESS in 4m 31s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/bc6040ec11194f8fbc7e62442779e942 : FAILURE in 26m 37s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/a14faddd75634033896f8e2ec8773689 : FAILURE in 49m 04s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/2d2f6d64459944faab99d5e3befed53a : FAILURE in 47m 17s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/941ac1cfef104bbb96f5317872f73b41 : FAILURE in 47m 29s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/a2a21cf9fb63410d9677dcfad534e2bc : FAILURE in 26m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6e8408443c03f2fd91f64ce515dd6477183caa63","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-05 01:43:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nThanks for the patch. One comment.","accounts_in_message":[],"_revision_number":2},{"id":"10ddb667a290d746a8fda351519bda271bc18e77","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-05 03:03:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3754bda04450099bc0ebc8f31266f074d6db0c3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 04:20:51.000000000","message":"Patch Set 3: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7722e123b13e431db83b31647b642fc5 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7584a0c95c4adda11da85c70e08486 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb1ed10863048b18f598added7cffb6 : FAILURE in 5m 24s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ecceae4d4b754a76a16b30e38455c4db : FAILURE in 50m 05s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/decc64a8766249dfbed7b25657ca7e99 : SUCCESS in 58m 15s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/79e73012d620490b8ce1099dfa4b8ee5 : SUCCESS in 46m 08s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/12c97e2d18f849f2be5ea91a170cd0d5 : TIMED_OUT in 1h 15m 05s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/42ecc38fe02245efbfb1c1b49eba24f6 : FAILURE in 31m 26s\n\nWarning:\n  Comments left for invalid file zun_tempest_plugin/tests/tempest/api/common/datagen.py","accounts_in_message":[],"_revision_number":3},{"id":"bd03dc057e314b30aa7386e16e83c614aeadc73b","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-05 04:31:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"68b84c30152c613d57d416bcde8040bfc65f089e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 05:31:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a64587819af44af6bd14a9fd71562b83 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f97c4242b4946e8aafb5bb4a83c6af0 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c416a79e5a8490e8107f1640c3bfbb3 : SUCCESS in 3m 32s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/28c1df6dcb934b1f97f2c4f10463200c : SUCCESS in 48m 18s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/03ae43e8c1914b1d96ab5e9dd823ab45 : SUCCESS in 58m 02s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/01ba856039754e96b035a87c1ec08069 : SUCCESS in 46m 45s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/ed28a5ff4bcc46b99abb8627bb4a6648 : SUCCESS in 46m 27s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/0c93b496f4a8418b916073afac02e802 : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":4},{"id":"ff2b25a6ad8b3e291f68d659dbe1dc548c30276d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-05 13:33:28.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bc5390a97c4de440d90aab18ba76212939482f0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 13:33:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b4ca71bd10e1ea277bb6d4d1fa45792099afa5cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 14:35:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a051e2d3aba64bdda0223415b153e925 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ab946e9c964abb9798ae6142d677f3 : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75611b51b58745d095196dad2f8b22cc : SUCCESS in 5m 47s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/fceb6b508a9c42639c196a5eb9352e09 : SUCCESS in 50m 54s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/32e398b04f1c4adb9109fdd523536472 : SUCCESS in 52m 07s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/b7eebefe56c6446583606c4c8ecc5356 : SUCCESS in 56m 28s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/017fd9f295a64bf29a84b47d8ff9e7d3 : SUCCESS in 47m 31s","accounts_in_message":[],"_revision_number":4},{"id":"8cfe42ce887a9cf8b8ec36db0c5d59a3e3040598","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 14:35:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"34b7723a3cf8be3f182a11bad4ef2a8d8c6764e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 14:37:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d63c424cee466aa81c1c4d329c65d2 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"06439d33a50b6c1ceb80ef2ae009575abb408e2c","revisions":{"c526774857e004e8c57c61149c98600d4d19e9e6":{"kind":"REWORK","_number":1,"created":"2020-05-04 06:44:55.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/92/725192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/92/725192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/1"}}},"commit":{"parents":[{"commit":"9043347888f26d5c9dc804f758597de2efea0181","subject":"Update docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9043347888f26d5c9dc804f758597de2efea0181"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-04 06:44:09.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-04 06:44:37.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nChange-Id: I1c9dd70d2d3319ad9012620fe8478316f6a940d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/c526774857e004e8c57c61149c98600d4d19e9e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/c526774857e004e8c57c61149c98600d4d19e9e6"}]},"branch":"refs/heads/master"},"6073bb37fe0b50fa3cb8108615a70092ab6e03fd":{"kind":"REWORK","_number":2,"created":"2020-05-04 07:50:25.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/92/725192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/92/725192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/2"}}},"commit":{"parents":[{"commit":"9043347888f26d5c9dc804f758597de2efea0181","subject":"Update docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9043347888f26d5c9dc804f758597de2efea0181"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-04 06:44:09.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-04 07:50:00.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nChange-Id: I1c9dd70d2d3319ad9012620fe8478316f6a940d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6073bb37fe0b50fa3cb8108615a70092ab6e03fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/6073bb37fe0b50fa3cb8108615a70092ab6e03fd"}]},"branch":"refs/heads/master"},"7fccb8eb41351c9ad2e2912eddd2624778dd054b":{"kind":"REWORK","_number":3,"created":"2020-05-05 03:03:43.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/92/725192/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/92/725192/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/3"}}},"commit":{"parents":[{"commit":"9043347888f26d5c9dc804f758597de2efea0181","subject":"Update docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9043347888f26d5c9dc804f758597de2efea0181"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-04 06:44:09.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-05 03:03:00.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nChange-Id: I1c9dd70d2d3319ad9012620fe8478316f6a940d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/7fccb8eb41351c9ad2e2912eddd2624778dd054b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/7fccb8eb41351c9ad2e2912eddd2624778dd054b"}]},"branch":"refs/heads/master"},"06439d33a50b6c1ceb80ef2ae009575abb408e2c":{"kind":"REWORK","_number":4,"created":"2020-05-05 04:31:07.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/92/725192/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/92/725192/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/92/725192/4"}}},"commit":{"parents":[{"commit":"9043347888f26d5c9dc804f758597de2efea0181","subject":"Update docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/9043347888f26d5c9dc804f758597de2efea0181"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-04 06:44:09.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-05 04:30:42.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nChange-Id: I1c9dd70d2d3319ad9012620fe8478316f6a940d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/06439d33a50b6c1ceb80ef2ae009575abb408e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/06439d33a50b6c1ceb80ef2ae009575abb408e2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
