)]}'
{"id":"openstack%2Fzun-tempest-plugin~725729","triplet_id":"openstack%2Fzun-tempest-plugin~master~Ia9ca8dc083354afb056880a00944a11782741bbe","project":"openstack/zun-tempest-plugin","branch":"master","topic":"stable-testing-tempest-plugins","hashtags":[],"change_id":"Ia9ca8dc083354afb056880a00944a11782741bbe","subject":"Add stable/ussuri job on master gate","status":"MERGED","created":"2020-05-05 23:02:06.000000000","updated":"2020-05-06 14:44:32.000000000","submitted":"2020-05-06 14:40:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725729-1588776038319-e143bcec","meta_rev_id":"7dd6751c7bdd2d755cdc8c06b74505a9478f4baa","_number":725729,"virtual_id_number":725729,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 14:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-05-06 13:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-05-06 13:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 13:31:17.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-05-06 14:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0873b6e7d3ef313d874506918686e22fe26513a3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 23:02:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d58f5be11de44723a9fbb53d6994a0163ef7990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 23:59:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af977116cc534e3f9b5081e5b170d899 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c829db72b54e9295c3edcc4b7831bb : SUCCESS in 3m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/ed0a46bdbe5648beb60c9d1f25d1c50e : SUCCESS in 42m 06s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/cd9f953126574098b3f187e3738640d8 : SUCCESS in 41m 41s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/e40ea52902f143dbb6c79e316ea45fdd : SUCCESS in 40m 37s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/b1420d37ec90457181852e3c3cba8f54 : SUCCESS in 45m 22s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/fe9b3aa96395413e86c35840e5970202 : FAILURE in 20m 38s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/ed8ff2e4cd574259b08cb452385b7c19 : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cbfd0abe2ae57c6627239b174fd0d72a9c7923d3","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-06 03:37:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"acb7dbe48f5a590f20713994cd6b35c71af296c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 04:37:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0091d3823f94c7fbb06ffc8705c7207 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/630d9a8a81984b88b9ee43fa751992b7 : SUCCESS in 3m 39s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/734347fcbe1a477bb99708f127788e07 : SUCCESS in 46m 48s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/36c854356c5740fbbb8b60b8b4825d2f : SUCCESS in 48m 16s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/20edc57c614d43cab8d6082ae2c6422b : SUCCESS in 50m 43s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/f9bf3958af8e4a5f84a9498ab08c6142 : SUCCESS in 44m 15s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/07b2b1aeeb1e434b9c50918820d38d04 : SUCCESS in 45m 37s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/789ed5c2823c49d7a837499fb839e0a5 : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d16545dd5561658a032f1650112ade77ada84bec","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-06 13:31:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"208699fea3e1724d270d3f0a4eb92671d07cf9c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:31:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7f8f54c7c86e4e7100f89115da67b462374b14f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 14:40:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efac11c9b8fe4db0a9b0a3ab51c2c91d : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/242eecfd65ac49ae9ee0e07c3f970fc2 : SUCCESS in 4m 02s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/17f5c403842942a2a4458c7edd0d0243 : SUCCESS in 56m 06s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/4d104520194c452ebe2f41509a2f1449 : SUCCESS in 51m 01s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/b3696635797243658d6dbe8e7434b9bb : SUCCESS in 49m 12s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/591bb71d712b4670ac943f7fe5c48152 : SUCCESS in 1h 00m 09s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/fc3f34b9657a41b4a80eeac7ef550f21 : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6f9abe8303ee5f0bb6dcfe3fc3f43d121c9ca4d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 14:40:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b7fc291c3e43ccd6b8036a60be4c3211d85b2aae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 14:44:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f8bac600a10415e9516fbcc34dc7372 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f2fd34ca8704e00b85bc5593fdc87aab62be4bf","revisions":{"7f2fd34ca8704e00b85bc5593fdc87aab62be4bf":{"kind":"REWORK","_number":1,"created":"2020-05-05 23:02:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/29/725729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/29/725729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/29/725729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/29/725729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/29/725729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/29/725729/1"}}},"commit":{"parents":[{"commit":"06439d33a50b6c1ceb80ef2ae009575abb408e2c","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/06439d33a50b6c1ceb80ef2ae009575abb408e2c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 22:47:07.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 23:00:42.000000000","tz":-300},"subject":"Add stable/ussuri job on master gate","message":"Add stable/ussuri job on master gate\n\nNow we have stable/ussuri branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the stable/ussuri job on tempest plugin\nmaster gate.\n\nChange-Id: Ia9ca8dc083354afb056880a00944a11782741bbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/7f2fd34ca8704e00b85bc5593fdc87aab62be4bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/7f2fd34ca8704e00b85bc5593fdc87aab62be4bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
