)]}'
{"id":"openstack%2Fzun-tempest-plugin~727612","triplet_id":"openstack%2Fzun-tempest-plugin~master~I80ac3e7869d315f23c66b3f9684a8b63209f0eb0","project":"openstack/zun-tempest-plugin","branch":"master","hashtags":[],"change_id":"I80ac3e7869d315f23c66b3f9684a8b63209f0eb0","subject":"Fix hacking min version to 3.0.1","status":"MERGED","created":"2020-05-13 02:44:28.000000000","updated":"2020-05-13 14:36:31.000000000","submitted":"2020-05-13 14:35:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727612-1589380502304-78abda74","meta_rev_id":"0f6248c85122874467bfb71eca9d08043b495c1c","_number":727612,"virtual_id_number":727612,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 14:35:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-05-13 13:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2020-05-13 13:21:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 13:21:47.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-05-13 14:35:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f660f05195e202976c720a4eef856245362d4472","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-13 02:44:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3980b200ef714fff18732d70d6d64a2cd9e7dc3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 03:49:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e585bfd28cad49ee817be7fb7d9b4a97 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/362e78449eba400bb3fa959e1b79b782 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5715926fb9f48d3b0f284bc55cf335c : SUCCESS in 6m 21s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/5a1635735a8d421b8ebcb6e9b04cbead : SUCCESS in 50m 47s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/d398d4249cf645da8aefffd9f7ebed0b : SUCCESS in 52m 15s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/bf91527269de4a58aae8f7fb43c07248 : SUCCESS in 48m 40s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/67a59ed9e49c455299dc2495942cca8c : SUCCESS in 49m 08s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/e30cb2ae3d19428bba361eb009f20921 : SUCCESS in 50m 34s\n- zun-tempest-multinode-docker-sql https://zuul.opendev.org/t/openstack/build/d4ebe9468cca401a951688406f52fa89 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2eb954731cbc0d021d5caf1c726496f58dfee555","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-05-13 13:21:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebbae0d6c936d2fd98fc40ee9f8fc7f986cb7d2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 13:22:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da6bcbd7954a76de46125855b028f358ece1967c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:35:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/04539400f02b43998c469edfe9262193 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec4b9c38c8a49c5b8f8a3b312498672 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6dc28a4a40e45dc8de13b862f6012a2 : SUCCESS in 4m 12s\n- zun-tempest-py3-docker-sql https://zuul.opendev.org/t/openstack/build/fd1bd22cd2dc4125a5f843cb85962bfe : SUCCESS in 49m 22s\n- zun-tempest-py35-docker-sql-rocky https://zuul.opendev.org/t/openstack/build/414ca93fc7434740b63bd70f06a1113a : SUCCESS in 46m 09s\n- zun-tempest-py35-docker-sql-stein https://zuul.opendev.org/t/openstack/build/aa0e17f17d874a02bccc0388d23cec07 : SUCCESS in 55m 56s\n- zun-tempest-py3-docker-sql-train https://zuul.opendev.org/t/openstack/build/fa82ec0dfced48eb9358e3e7d85d38bf : SUCCESS in 54m 08s\n- zun-tempest-py3-docker-sql-ussuri https://zuul.opendev.org/t/openstack/build/1383a8265add4151a44ed32c75c795a9 : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":1},{"id":"cdabc007126f31b3ab17a17664da24fa220e05bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:35:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0f6248c85122874467bfb71eca9d08043b495c1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:36:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe3b26405db43a0a3665054fa526615 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5df462dcfa971b743b1305cbf5d44bec585ee436","revisions":{"5df462dcfa971b743b1305cbf5d44bec585ee436":{"kind":"REWORK","_number":1,"created":"2020-05-13 02:44:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/727612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/zun-tempest-plugin","ref":"refs/changes/12/727612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/12/727612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/12/727612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/12/727612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/zun-tempest-plugin refs/changes/12/727612/1"}}},"commit":{"parents":[{"commit":"552edb7146cc262e4a27932098e9c884804eb7a2","subject":"Merge \"Add stable/ussuri job on master gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/552edb7146cc262e4a27932098e9c884804eb7a2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 02:43:36.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 02:44:27.000000000","tz":-300},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: I80ac3e7869d315f23c66b3f9684a8b63209f0eb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5df462dcfa971b743b1305cbf5d44bec585ee436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/zun-tempest-plugin/commit/5df462dcfa971b743b1305cbf5d44bec585ee436"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
